)]}'
{"id":"openstack%2Fpython-manilaclient~982605","triplet_id":"openstack%2Fpython-manilaclient~master~I91fd0f94a24119a378fa2660260719fc37290c82","project":"openstack/python-manilaclient","branch":"master","attention_set":{},"removed_from_attention_set":{"38967":{"account":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"last_update":"2026-04-12 13:35:34.000000000","reason":"\u003cGERRIT_ACCOUNT_38967\u003e replied on the change","reason_account":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"}}},"hashtags":["outreachy-2026-prospect"],"change_id":"I91fd0f94a24119a378fa2660260719fc37290c82","subject":"Fix copy-paste errors in share_networks and   share_network_subnets docstrings","status":"ABANDONED","created":"2026-03-30 02:29:02.000000000","updated":"2026-08-24 05:44:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca4009cf2fbb54f68891c4b26089a813d8ce6979","_number":982605,"virtual_id_number":982605,"owner":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-30 09:56:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2026-04-02 06:26:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 03:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 06:26:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"b5228c932866d119f1b2071c8c806b281287ca61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-03-30 02:29:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0c847004ec15b71a9602df63489343dfb05304f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 03:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638b38cc66d148f8bf8473e2370a4fbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afa81be441d469ab23588383fd1ca9f : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b9039ce9f4448c8c0ea7c1ddba21ad : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e02d18a0a6421d8b7445f04dae2c60 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/117c7535bf4142f3b6d8bb05a3a344f3 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbc9255afc3a41d698c2ed270cdc2401 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fdf8ffdeae954ad6b00a6ab0dac889c4 : SUCCESS in 8m 07s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/aec54c0bb1634fdb8de390ea948184ff : SUCCESS in 8m 51s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/d732832a19774878b999bf59f4e8fbec : SUCCESS in 30m 53s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/d175892bf73044628b4fc159f75de268 : SUCCESS in 4m 40s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f5c3e93b639e4599a30f61b4a9581019 : FAILURE in 34m 58s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/f936fbf37ce64b6babfb1cc0972b8639 : POST_FAILURE in 13m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/14febf067bfb4888ba2872725d4dbcb3 : FAILURE in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61521ad0e14e1616e54f1654d35437dd7a219fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-03-30 09:14:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae3cf8a2be4059e12e912861a402c925dcd1daf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 09:56:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95970bd68b8d4479bc1d85b2b4ddfd18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8cc53852ed4c9490b5958b1370d766 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/602faca5729d4f30ab084bca7fde2b55 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6101d2f76d2b411ba72b06972a1d908b : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a58ccdaac244849aa57b35035690630 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fd263a48e1e47e7a2e9138e8db9bc62 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4ffcf6173be4b138cbc3b3650d6619e : SUCCESS in 5m 07s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ae8e1435b00240c99ca27fcc84ba722e : SUCCESS in 5m 51s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/5c1bf1076fd74fe98b4b42e94596ddb7 : SUCCESS in 32m 31s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/4e926a8a572449268f7559f5b7eb4a0b : SUCCESS in 3m 05s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c1667c3726e94443ac2e190cf3902c37 : FAILURE in 31m 21s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/014192bc7b5f4caaa3ee666f96d5b819 : POST_FAILURE in 21m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/23f99377cc0047dab1cb3df57f8fcd90 : FAILURE in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4b8915beeda1066a42815e52e6705327b2d1fba","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-02 06:00:00.000000000","message":"Hashtag added: outreachy-2026-prospect","accounts_in_message":[],"_revision_number":2},{"id":"a7563b4e1d6c74722c9f761e160aa183d4ceafb5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-02 06:26:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6b5f87654ac5e47954c44284278e0ecc7e1286c","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-12 13:35:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca4009cf2fbb54f68891c4b26089a813d8ce6979","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-24 05:44:14.000000000","message":"Abandoned\n\nalready done","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e042a575721b84ff330f41572ec753b16609fa3","revisions":{"a4d9148dda1f081a2c3f96d9dd0036faff18a65f":{"kind":"REWORK","_number":1,"created":"2026-03-30 02:29:02.000000000","uploader":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"ref":"refs/changes/05/982605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/05/982605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/1"}}},"commit":{"parents":[{"commit":"6c2b5c808530bf2c8eb467adfd272ba010224f10","subject":"Fix typo \u0027occured\u0027 should be \u0027occurred\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/6c2b5c808530bf2c8eb467adfd272ba010224f10"}]}],"author":{"name":"Insitupeter","email":"insitupeter@gmail.com","date":"2026-03-30 02:16:41.000000000","tz":0},"committer":{"name":"Insitupeter","email":"insitupeter@gmail.com","date":"2026-03-30 02:28:32.000000000","tz":0},"subject":"Fix copy-paste errors in share_networks and   share_network_subnets docstrings","message":"Fix copy-paste errors in share_networks and\n  share_network_subnets docstrings\n\nChange-Id: I91fd0f94a24119a378fa2660260719fc37290c82\nSigned-off-by: Insitupeter \u003cinsitupeter@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/a4d9148dda1f081a2c3f96d9dd0036faff18a65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/a4d9148dda1f081a2c3f96d9dd0036faff18a65f"}]},"branch":"refs/heads/master"},"5e042a575721b84ff330f41572ec753b16609fa3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-30 09:14:33.000000000","uploader":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"ref":"refs/changes/05/982605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/05/982605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/05/982605/2"}}},"commit":{"parents":[{"commit":"6c2b5c808530bf2c8eb467adfd272ba010224f10","subject":"Fix typo \u0027occured\u0027 should be \u0027occurred\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/6c2b5c808530bf2c8eb467adfd272ba010224f10"}]}],"author":{"name":"Insitupeter","email":"insitupeter@gmail.com","date":"2026-03-30 02:16:41.000000000","tz":0},"committer":{"name":"Fiyin Peter","email":"insitupeter@gmail.com","date":"2026-03-30 09:14:33.000000000","tz":0},"subject":"Fix copy-paste errors in share_networks and   share_network_subnets docstrings","message":"Fix copy-paste errors in share_networks and\n  share_network_subnets docstrings\n\nCloses-Bug:#2146281\nChange-Id: I91fd0f94a24119a378fa2660260719fc37290c82\nSigned-off-by: Insitupeter \u003cinsitupeter@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/5e042a575721b84ff330f41572ec753b16609fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/5e042a575721b84ff330f41572ec753b16609fa3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
