)]}'
{"id":"openstack%2Fpython-manilaclient~982695","triplet_id":"openstack%2Fpython-manilaclient~master~Ic2da3173ae33142335a815557526283f81bbdd4c","project":"openstack/python-manilaclient","branch":"master","topic":"bug/2146755","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-08-24 06:03:41.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-02 06:28:17.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38842":{"account":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"last_update":"2026-08-24 06:03:41.000000000","reason":"Change was submitted"}},"hashtags":["outreachy-2026-prospect"],"change_id":"Ic2da3173ae33142335a815557526283f81bbdd4c","subject":"Mock sleep in ClientTest to reduce unit test runtime","status":"MERGED","created":"2026-03-30 17:01:15.000000000","updated":"2026-08-24 06:06:03.000000000","submitted":"2026-08-24 06:03:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982695-bug/2146755","meta_rev_id":"b671ec519b6850496a4ebd74084194e80f270ed8","_number":982695,"virtual_id_number":982695,"owner":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 06:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-04-02 06:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-08-24 05:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 17:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 06:42:20.000000000","updated_by":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-31 06:42:20.000000000","updated_by":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"32c291c08f50cf9048a8bf5912c0d5ce68f24e59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-03-30 17:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"328fa1169628ee6520d3594ab1cbc5a4cc725fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:30:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0433f4b8cd3041e28cab23a309f8acaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483dda1559d84fabbf756ac60d2922b1 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dead925aa4174acab686d19e2469fd74 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d353f38a2d440b6932b09028abdc2b4 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca5331656b034c0aa5415a3bffac6277 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f82b686be584820bb547068e0dfe469 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99d85872fe4e494782e33cd9f9921339 : SUCCESS in 7m 07s (non-voting)\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/fd6c9c489c87488c9b73c4e8338b7645 : SUCCESS in 3m 10s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ac0c6345f57b417ca7a7ff3528556e88 : FAILURE in 21m 25s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a32e1156737d456d85bb1a62e09f8227 : FAILURE in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f9e1c38dc0b9bf8c6aca15e7d5e6e05e6bcd10e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-31 20:05:59.000000000","message":"Hashtag added: outreachy-2026-prospect","accounts_in_message":[],"_revision_number":1},{"id":"037de1378d45b6b640f056a8966005b50f2bdd5c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-02 06:28:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d190650b43b4d734797bb8af5935479a938da3f7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-02 06:28:26.000000000","message":"Topic set to bug/2146755","accounts_in_message":[],"_revision_number":1},{"id":"4b2e60e075d51c9e694f591da96e3deace62b04d","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-04-13 12:00:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2afd410bf46f292190ce783bbec1666754a954c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-24 05:42:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70c75fc5d5a434a6366e36c83a656393398dfb49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 05:43:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76a68910bda812b9ca155eaa2e05a01298c2cdca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:03:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c5c123a67e438a877c7ea053fee8eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df49a0db0324b7ba0268dc226215dab : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb66fd8da6b0451f8ea767dc6928b1ed : SUCCESS in 2m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26e8f942a22d4315b6d66284cdee6137 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c003f13e65c48cbacc02aaa63bf63a9 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a13bf43ea2ca46058b8f895a750e2d0b : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":1},{"id":"52a54ceae418531d0e0c57a9997450631ec6aa01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:03:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b671ec519b6850496a4ebd74084194e80f270ed8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:06:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f0c813bdced4dbbacb576ea5f64cb75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bccdbd4329f406780ae9df8bb4039df : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb40b7f777a593bef72d2016ee02ce888d7d05dc","revisions":{"fb40b7f777a593bef72d2016ee02ce888d7d05dc":{"kind":"REWORK","_number":1,"created":"2026-03-30 17:01:15.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/95/982695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/95/982695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/95/982695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/95/982695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/95/982695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/95/982695/1"}}},"commit":{"parents":[{"commit":"6c2b5c808530bf2c8eb467adfd272ba010224f10","subject":"Fix typo \u0027occured\u0027 should be \u0027occurred\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/6c2b5c808530bf2c8eb467adfd272ba010224f10"}]}],"author":{"name":"Silvia-Wachira","email":"wachirasilvia8@gmail.com","date":"2026-03-30 16:41:45.000000000","tz":180},"committer":{"name":"Silvia-Wachira","email":"wachirasilvia8@gmail.com","date":"2026-03-30 17:00:54.000000000","tz":180},"subject":"Mock sleep in ClientTest to reduce unit test runtime","message":"Mock sleep in ClientTest to reduce unit test runtime\n\nThree unit tests in test_httpclient.py were taking approximately\n1 second each due to a real sleep() call in the retry logic of\nmanilaclient/common/httpclient.py.\n\nThis change introduces a mock for sleep in the ClientTest setUp()\nto eliminate real waiting time without altering test logic.\n\nAffected tests:\n- ClientTest.test_retry_limit          (1.007s -\u003e 0.005s)\n- ClientTest.test_get_retry_500        (1.003s -\u003e 0.003s)\n- ClientTest.test_get_retry_400_socket (1.003s -\u003e 0.004s)\n\nCloses-Bug: #2146755\nChange-Id: Ic2da3173ae33142335a815557526283f81bbdd4c\nSigned-off-by: Silvia-Wachira \u003cwachirasilvia8@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/fb40b7f777a593bef72d2016ee02ce888d7d05dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/fb40b7f777a593bef72d2016ee02ce888d7d05dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
