)]}'
{"id":"openstack%2Fpython-manilaclient~983388","triplet_id":"openstack%2Fpython-manilaclient~stable%2F2025.1~I2d971f2c4771d62a1776c995ef8a16594fd04c99","project":"openstack/python-manilaclient","branch":"stable/2025.1","topic":"bug/2057951","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-07-31 16:49:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d971f2c4771d62a1776c995ef8a16594fd04c99","subject":"fix respect region name in client initialization","status":"MERGED","created":"2026-04-05 18:10:45.000000000","updated":"2026-07-31 16:50:48.000000000","submitted":"2026-07-31 16:49:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983388-bug/2057951","cherry_pick_of_change":967138,"cherry_pick_of_patch_set":1,"meta_rev_id":"98b4d353c825d96a10bed864b13dbd01a1ce8317","_number":983388,"virtual_id_number":983388,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 16:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-07-31 15:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-07-31 15:57:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 18:10:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"state":"CC"},{"updated":"2026-04-05 18:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-31 12:55:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-07-31 15:57:33.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"d609865846f81e10c5d1ead4d2ff56f976c126fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:10:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"7fff20f867c7887e467668e349c94ede13b717c2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:11:05.000000000","message":"Topic bug/2057951-stable/2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"ff8b392a580927f7d3372d41890473a4b3001798","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:11:08.000000000","message":"Topic set to bug/2057951","accounts_in_message":[],"_revision_number":1},{"id":"e807c493fd65a7cacbaf2d83d7cd93887f054f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 18:54:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74e2bf48da4b4fdd8f28834d3b472ac9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60872d6088342568438f7d310b7b81e : FAILURE in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fefe2b0e71dd466b847025da3db84cae : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a522e389e1048d1ae3482bdd39b904e : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf7f678ad1ac44f4a4e2d8058e33446e : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a254aca95c94ca288536c222cdf0269 : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db3fcd4ef95541a4a0c17552fe4e7e25 : SUCCESS in 4m 30s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/cfaffcc8dfe94d8f9c5a1d4aabe47e98 : FAILURE in 23m 52s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/18656ba619ff4c3593bba928ab616746 : SUCCESS in 4m 02s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/977264e6e35e4811bded13514f014b2c : SUCCESS in 36m 22s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/72039db6fe12416baa8ac84e47906eb3 : SUCCESS in 43m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5a1bee036ac246e39e003c061f352752 : SUCCESS in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a456dc77d902c26ea2693dd9cef8c09f39658ff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-19 05:47:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53a3c689f324b874f887b5fdbc40ee0ac18619c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 06:53:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e8f54a69c994265b1d4779e89ab49c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c75d811c321473c8c4ee52aa3dad55c : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/637a71920f394366b8087b801b0352e8 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8dee0629a1341b9ace3175efbb8aa90 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95049328f6504cada9fc4d6a88222fdb : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ddebc8130f54e29af4f258b66a6ff82 : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f38c9ee47d2744bd9d31edf9b86c7e4f : SUCCESS in 4m 00s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/f3ae1165bd79461cbc210a639ea55eb7 : FAILURE in 21m 54s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/46f4022e6c624a0bbd3ccccd5e9e4b41 : SUCCESS in 3m 00s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/44087389aae2418cbf381d76effb0255 : SUCCESS in 33m 10s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/58d62a97adec45afb1dfd936f7daae8a : SUCCESS in 58m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ed44b195985a410e9e05ee57a747d1a7 : SUCCESS in 31m 19s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"db1492210488033566430621c4049c819adf7459","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-31 12:55:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90a5fcc4468d697a87c045bde10b60dbb9528f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 14:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18bc93927bf84eac9fd8c96c69cba992\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6c9053b5214d0493342bcfa7ddb2d5 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/552c2300b6ba4a579a63187fa0b68bb3 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5acdd25510845a9a121e7c63a4b232f : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5b73bde0dd840718041224b0d3b8511 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1da02648ef384765b7483169c05f8181 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/285c12ab68074a5c8da48a828df1068a : SUCCESS in 5m 13s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/ddd7feff27a340f09323aa66a33e88a5 : SUCCESS in 1h 07m 27s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/c76f54613b2747a9b4d42842790cf8b9 : SUCCESS in 2m 50s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a58fdb6bddbf48e6aeabb32ddd352f86 : SUCCESS in 37m 14s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/7156dbc698654c04a61ebb50373c2c03 : FAILURE in 19m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/94c2cca96954470c995d2ee7b7226a1d : SUCCESS in 32m 39s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"d29c2fdea49564da74674530a7513af7f542d3ae","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-31 15:57:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89bf286debcff77ec8d540751711e63518ca4fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 15:57:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"979a1eed3e2371ff69727b22cee9571566616e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 16:49:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f4f40bc686464da351c6c648c7c1ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa491566fb3401c9ae01f0421c30b54 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e7c6412726407c92715bd0ca84aa44 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12752a9ba9a442888592bf131daab4ec : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/420e0fe27534409a9fe4f1844332522c : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6ce95193f144f81b4f040cdd71bb007 : SUCCESS in 4m 16s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/7ba9a5cc01ca4df8a9a78c0e3d4341b4 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3c9dc14e9d6530e553c95ca17ae7ef14d3b7fcb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 16:49:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98b4d353c825d96a10bed864b13dbd01a1ce8317","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 16:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b5ef56f944452a810d104b0a0cab3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d07fca287c843438f95c3af4ecd2f86 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d33f0fbd3f9e41aba9d554dee0602ea2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1449264426f7a1ad8e44b97ad4eb209a2874e57","revisions":{"d1449264426f7a1ad8e44b97ad4eb209a2874e57":{"kind":"REWORK","_number":1,"created":"2026-04-05 18:10:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/983388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/88/983388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/88/983388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/88/983388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/88/983388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/88/983388/1"}}},"commit":{"parents":[{"commit":"4a1177ffb19c76b496aeba68ddf0f8739ad64019","subject":"Fixes an issue that prevented ensure shares to be run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/4a1177ffb19c76b496aeba68ddf0f8739ad64019"}]}],"author":{"name":"denver-baraka","email":"denverbaraka@gmail.com","date":"2025-10-23 17:26:59.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-05 18:10:45.000000000","tz":0},"subject":"fix respect region name in client initialization","message":"fix respect region name in client initialization\n\nwhen initializing manila python client, the region\nname was being ignored. As a result, the client\ncould connect to the wrong region endpoint.\n\nThis change ensures that region name is passed to\nsession.get_endpoint() so the correct endpoint is\nused.\n\nTest,\n-Updated unit tests to verify that the region name\n is respected during client initialization.\n\nCloses-Bug: #2057951\nChange-Id: I2d971f2c4771d62a1776c995ef8a16594fd04c99\nSigned-off-by: denver-baraka \u003cdenverbaraka@gmail.com\u003e\nAssisted-By: Copilot\n(cherry picked from commit 8019cc5ac0850e00ff8d52a82a627aa4a97ab11c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/d1449264426f7a1ad8e44b97ad4eb209a2874e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/d1449264426f7a1ad8e44b97ad4eb209a2874e57"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
