)]}'
{"id":"openstack%2Fpython-manilaclient~983389","triplet_id":"openstack%2Fpython-manilaclient~stable%2F2025.2~Ic196c7bf8326b1e97d51241fbf0e30342f5e437b","project":"openstack/python-manilaclient","branch":"stable/2025.2","topic":"bug/2132937","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-07-31 19:11:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic196c7bf8326b1e97d51241fbf0e30342f5e437b","subject":"Add missing api_versions decorator for _update_share_group","status":"MERGED","created":"2026-04-05 18:11:34.000000000","updated":"2026-07-31 19:12:15.000000000","submitted":"2026-07-31 19:11:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983389-bug/2132937","cherry_pick_of_change":968436,"cherry_pick_of_patch_set":2,"meta_rev_id":"7c7f140ae38fb3678ea950aa21b58a863e558d5e","_number":983389,"virtual_id_number":983389,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 19:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-07-31 18:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-07-31 18:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-05 18:11:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"CC"},{"updated":"2026-04-05 18:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-30 21:22:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-07-31 18:29:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"605b7c568fd1a035ca3a86ffa6c87acb9cd82e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:11:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8bea9f1a88ac403001d1826322904a4839596780","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:11:51.000000000","message":"Topic bug/2132937-stable/2025.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"85f5097d662a446a3a5fe7c3378657449d997028","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-05 18:11:56.000000000","message":"Topic set to bug/2132937","accounts_in_message":[],"_revision_number":1},{"id":"34cdaf409bbff70a555f2ad85b1dc6986aef8ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 18:50:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72ffc26a51b34df39736b525c60e2120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78ea11762554a50bd8a57b8c18d561a : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4199226e727e4a22a9851fce0609e7b3 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49204d04949c486db256393770a9318b : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80ca18f1e7424d29ba0286f6264e2f1f : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19beef010106424ca70d0a3bc0108cad : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b7a58f81f094ab3af85ae24e997c5d0 : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2fd122002ae4521ac9dd06e4239da29 : SUCCESS in 4m 10s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/0cd348630f7b49459f655c1af4116d8c : FAILURE in 13m 49s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/decd3284c4ba4c56a5f8e5d0a7018587 : SUCCESS in 3m 11s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5055e409d0414c34b191fbbfee7cf089 : SUCCESS in 33m 23s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/f5aac35bb6834f16a2ab17a3fb13150c : POST_FAILURE in 9m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/35caa81729944eea8641490d16db125b : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5608976589f6300e314471d23e10330d405fbeb7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-30 21:22:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e030d9a91b79b512629a578900d23c34c0641a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 22:24:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d57224678c545488c39b36cc7323d36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80bcfae4d6e644f8b0ba6c73acbfc175 : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed59d1c9e7da4b2c93976a076c48b36a : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4262c79342514aa3972c066e06f1a6b0 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb54ff370bed48d49d02cbd8ab97d0a0 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd7d8a0a1d2a49ec90fec24381e2b268 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa57ce09122a404aaa2172e3a202316a : SUCCESS in 6m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ccbc908b21a4bda9ea6c00f5132b02d : SUCCESS in 2m 45s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/2875600133b4495ea934233130f2a245 : SUCCESS in 57m 33s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/a899662320d449b98ee2830b88028c80 : SUCCESS in 2m 52s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5256a81a8fbc4e97953a64eeeb0f854f : SUCCESS in 32m 35s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/3fdeeb0b69e747ba859325f6a48ff003 : POST_FAILURE in 20m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6e1ec8958ba34bd2a03210e369334621 : SUCCESS in 16m 18s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"65268aaed494372ec16db581e0fd9e5bc6a2cac9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-31 18:29:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8504aa22bb061bca6a7eec197666d53c89b33ad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 18:30:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e15c30e22b222e8f1c59a017973e6c9fcdc31f0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:11:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9e40516ca54162b1b83fa7e0857ad2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3374ad0a51d4fe089b4e30f3ed52db6 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02da18e05dbd4ae8a2dece1eab7f39a6 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc41433c808f43d985a945449d2affbe : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccb748889c6540f597127ef2145ab63e : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/916076afdbf84ecb9b42611399155fd0 : SUCCESS in 4m 12s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/cb45b46fd46144699c235201282e0171 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"873412229ccbc6573f240d36783299873bed2b74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:11:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c7f140ae38fb3678ea950aa21b58a863e558d5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:12:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c81f7c883145e78109c36937aa8381\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2a150d5fd84a049c04d8badf499b66 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e411852dd1b40269f170ddc35325100 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58b62ee683e4eb3b43e84521f919d4986e002eae","revisions":{"58b62ee683e4eb3b43e84521f919d4986e002eae":{"kind":"REWORK","_number":1,"created":"2026-04-05 18:11:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/983389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/89/983389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/89/983389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/89/983389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/89/983389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/89/983389/1"}}},"commit":{"parents":[{"commit":"b44aae26b80e21db3f3c803ea5259fa7f58573e4","subject":"Fixes an issue that prevented ensure shares to be run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/b44aae26b80e21db3f3c803ea5259fa7f58573e4"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-11-26 09:15:28.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-05 18:11:34.000000000","tz":0},"subject":"Add missing api_versions decorator for _update_share_group","message":"Add missing api_versions decorator for _update_share_group\n\nAdd missing api_versions.experimental_api decorator for\n_update_share_group\n\nCloses-Bug: 2132937\nChange-Id: Ic196c7bf8326b1e97d51241fbf0e30342f5e437b\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n(cherry picked from commit 875a29ed93895efc5ad2ecc296fa3c5536fa3ee8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/58b62ee683e4eb3b43e84521f919d4986e002eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/58b62ee683e4eb3b43e84521f919d4986e002eae"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
