)]}'
{"id":"openstack%2Fpython-manilaclient~984064","triplet_id":"openstack%2Fpython-manilaclient~master~I813c6715add91cf46c3600b8793e018368d227e0","project":"openstack/python-manilaclient","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-24 05:46:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"38967":{"account":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"last_update":"2026-08-24 06:25:00.000000000","reason":"Change was submitted"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2026-08-24 06:25:00.000000000","reason":"Change was submitted"},"38842":{"account":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"last_update":"2026-04-10 21:03:11.000000000","reason":"\u003cGERRIT_ACCOUNT_38842\u003e replied on the change","reason_account":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"}}},"hashtags":[],"change_id":"I813c6715add91cf46c3600b8793e018368d227e0","subject":"Add missing share backup and share transfer sections to CLI docs","status":"MERGED","created":"2026-04-10 16:12:09.000000000","updated":"2026-08-24 06:26:37.000000000","submitted":"2026-08-24 06:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984064","meta_rev_id":"9da95b654a613705278f189187abde0f50d78a89","_number":984064,"virtual_id_number":984064,"owner":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 06:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-08-24 05:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-08-24 05:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"reviewers":{"CC":[{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 16:13:32.000000000","updated_by":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-04-10 16:13:32.000000000","updated_by":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2026-04-10 16:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-10 18:53:20.000000000","updated_by":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"reviewer":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"state":"CC"}],"messages":[{"id":"e696eacfa3282830b06ef9c373079321985d9561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-10 16:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45e2fa24fbe83687e7995c880fb73b0c937f3bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 16:51:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8f641bf88a445398bbe921524f1ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086fc73fce4040138ce44622ec91b69f : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2cf968b35f849ec994061ca9af85d20 : SUCCESS in 3m 18s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/ce11a45fd74a48ad91bc116d11c63699 : SUCCESS in 33m 58s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/23e7c11435c44524967373bc5e7ec1fa : SUCCESS in 2m 40s\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/85e221606b2a45baa7a859dea3a2909f : POST_FAILURE in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58ce58ccc76f197749efa8ab31b44aae2a7ce83f","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-04-10 18:53:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49859adf8d091ac0e9eb0b6bd325cc48c0e08615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-10 20:41:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"874ef7d7fb5a7d17b738b64f76828605fa7c2dc5","author":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"date":"2026-04-10 21:00:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fef7f3b562231574e363fae4ec3489910e75998d","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-04-10 21:03:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"431945285f7c3add83a414a4f9717aed869640fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 21:07:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33bdf7907e344217ba395d2c000e9330\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd33bfbb06f4568adea5cd011482549 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25659e9fdb2d4b02a5b412a5661fedb4 : SUCCESS in 3m 26s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/43b15c54de3f428e87cb237fcb648c8c : SUCCESS in 23m 22s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/fad66cb0e7a7458eaee4c9aed5af707e : SUCCESS in 3m 00s\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/d9078872de584483a9c11456314bdfd6 : POST_FAILURE in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9168d3c731916fcbf089e6b3a8c6d0d1b1c4ce3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-24 05:46:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"812f28b3d0d72f169d4a274d142af4e6c67503cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 05:47:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b78935323493f2e14de93305f100e00148744c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:25:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d46fc8c5612475ab0723c2bc7df6ab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e698a53a0134f8d912990eff626673f : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa1c3c3430047479e05daa8e65328ec : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eef80263c0c24491917b5fbe08bedb84 : SUCCESS in 2m 35s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/6c44a2c8844e47308c68680a53c68bb1 : SUCCESS in 23m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a3fb28480018b00669edc4f1e7d32487645e3a37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:25:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9da95b654a613705278f189187abde0f50d78a89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:26:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85732b8c9dae4749b55b948fbf0b205a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3a59126fef4414a06232e47e82c896 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"64af6a51f27df3d653b24e04076adeb590a72d41","revisions":{"44b8951727d59fd35ef2b668834850d44a2055b4":{"kind":"REWORK","_number":1,"created":"2026-04-10 16:12:09.000000000","uploader":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"ref":"refs/changes/64/984064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/64/984064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/1"}}},"commit":{"parents":[{"commit":"6fb890908a4bf0e5b7fdc15b4fbeb0d0dbf4767e","subject":"Merge \"update stale OpenStack references in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/6fb890908a4bf0e5b7fdc15b4fbeb0d0dbf4767e"}]}],"author":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-10 16:07:50.000000000","tz":0},"committer":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-10 16:11:36.000000000","tz":0},"subject":"Add missing share backup and share transfer sections to CLI docs","message":"Add missing share backup and share transfer sections to CLI docs\n\nThe share backup and share transfer command groups were missing\nfrom the CLI command reference at doc/source/cli/osc/v2/index.rst\ndespite being fully implemented and registered as entry points.\n\nCloses-Bug: #2147984\nChange-Id: I813c6715add91cf46c3600b8793e018368d227e0\nSigned-off-by: Insitu \u003cinsitupeter@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/44b8951727d59fd35ef2b668834850d44a2055b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/44b8951727d59fd35ef2b668834850d44a2055b4"}]},"branch":"refs/heads/master"},"64af6a51f27df3d653b24e04076adeb590a72d41":{"kind":"REWORK","_number":2,"created":"2026-04-10 20:41:54.000000000","uploader":{"_account_id":38967,"name":"Fiyin Peter","display_name":"Fiyin Peter","email":"insitupeter@gmail.com","username":"kitana"},"ref":"refs/changes/64/984064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/64/984064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/64/984064/2"}}},"commit":{"parents":[{"commit":"6fb890908a4bf0e5b7fdc15b4fbeb0d0dbf4767e","subject":"Merge \"update stale OpenStack references in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/6fb890908a4bf0e5b7fdc15b4fbeb0d0dbf4767e"}]}],"author":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-10 16:07:50.000000000","tz":0},"committer":{"name":"Insitu","email":"insitupeter@gmail.com","date":"2026-04-10 20:34:51.000000000","tz":0},"subject":"Add missing share backup and share transfer sections to CLI docs","message":"Add missing share backup and share transfer sections to CLI docs\n\nThe share backup and share transfer command groups were missing\nfrom the CLI command reference at doc/source/cli/osc/v2/index.rst\ndespite being fully implemented and registered as entry points.\n\nCloses-Bug: #2147984\nChange-Id: I813c6715add91cf46c3600b8793e018368d227e0\nSigned-off-by: Insitu \u003cinsitupeter@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/64af6a51f27df3d653b24e04076adeb590a72d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/64af6a51f27df3d653b24e04076adeb590a72d41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
