)]}'
{"id":"openstack%2Fpython-manilaclient~998703","triplet_id":"openstack%2Fpython-manilaclient~master~I8dc604d83425f16aeb033d893336daea06915e52","project":"openstack/python-manilaclient","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-08-27 11:41:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-27 02:33:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36686":{"account":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"last_update":"2026-08-27 12:20:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8dc604d83425f16aeb033d893336daea06915e52","subject":"Add availability_zone filter to share list (2.97)","status":"MERGED","created":"2026-07-25 21:55:02.000000000","updated":"2026-08-27 12:22:52.000000000","submitted":"2026-08-27 12:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998703","meta_rev_id":"f0505b1628659aa35102182198d20c0910e316a1","_number":998703,"virtual_id_number":998703,"owner":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 12:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-08-27 11:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-08-27 02:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-08-27 11:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-25 22:35:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 05:26:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-08-26 22:27:46.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-08-27 11:41:37.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"e142e9aece78dc4c4ca2dc98ef0d3414b0a37fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-07-25 21:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"575008d1ed3e2ebb1e60e4e84fb623237743c5ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 22:35:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9c747c7649495ba05e0a2681a0383a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2941111edd4d4637bd08ae5b142b3de1 : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ef1cfd8128f4f8f949e9bcbc5505cd5 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a98f1ef659ea440da60e7b9d5964236a : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/799efc51c9024830a71c078a65e1ac37 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcb7e25bd5424ca88501537904f6fc7c : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f210b34cec9641c19f6cdc42109fb038 : SUCCESS in 4m 27s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/96490571e2c348f9963aae009d4053c5 : SUCCESS in 5m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182418c6ea3c42b998657501913df65e : SUCCESS in 2m 05s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/3224ee38f0274801afaef619888c01bc : SUCCESS in 22m 18s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/c2747d6a65434468adecb9546b737925 : SUCCESS in 6m 04s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4edf9738e6114421b7b02360bc0f4682 : SUCCESS in 23m 44s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/a7b1116cbca84a4cb2c0ca96923a873d : POST_FAILURE in 16m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f6a6190e22e74a8f9f31aeea44c0622c : SUCCESS in 34m 53s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"745e2ff1f8e91e645b9f91b2666e8df47cd067cf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-24 05:26:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45ddcf4a5c560b5da174ab6fe0bbb1ccaa27d739","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-26 22:27:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66a9c9b5facb944b2b4f56a903f8501cb33005d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-08-27 00:40:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"34236acd5bb7540a6df85c8c1f54c52567aa5e1e","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-08-27 00:44:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ea433e59b5902f0406f33f564f0ad211daaf0147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 01:20:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30148152fac941368184ff6c89083fba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d24b0bb9f9fb4077b876288df5730ca3 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80db461080e84578b2ed6e8fe93e88f0 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24343cd504a4090a16a4a8ccd279b3a : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c7d928a17e144ce9a882ca740f6b93f : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e99cb9a078e54c7fbcec0248b152c3df : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/544a14b844c0442b8f65bcb27932a3ba : SUCCESS in 7m 39s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/255d825d8b0d49af9a575c3061b7c9a9 : SUCCESS in 5m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a4ae193a72f41e0817a2781715ba88b : SUCCESS in 3m 50s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/06d0665fe11247faa48d97e0d27a55da : SUCCESS in 21m 29s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/c169ee1e76954b3c8b7794ec54078e58 : SUCCESS in 1m 56s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/343532e58f7c4fd8901f9696103c9214 : SUCCESS in 17m 31s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/2ed85b45a515406aa48086f5135f0437 : POST_FAILURE in 9m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b7d4aaa9ae543c7b5e0c061fbbf04db : SUCCESS in 34m 12s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"1fa9b5719528946bf8bc8b6175d194a3b5063ae5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-27 02:33:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10fc3aff45f76029d3d8ccd5643e5722930d2356","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-27 11:41:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9419e3515108fea61bc24a73d0a5377d1a420ee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:42:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"776b964616af53a6310338fac87d93aef9116d64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:20:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f5b967bf10418c81cefa113ba4b10b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d6edf5f5f94d1abd90e595a0c62408 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f43493b3a54f5fb5c3193199eee722 : SUCCESS in 7m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26c0b502c274444c906853300b8e04d2 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baaa8a169ca745c4a7d1a87cd172905f : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/631a9cec213142508c9e2e08c5e94acf : SUCCESS in 2m 41s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/f4956d5630ac4db5b1e30766196407fa : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2be25e1f4ff2f16fab9ddc7471998bd6565b9f5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f0505b1628659aa35102182198d20c0910e316a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 12:22:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc80849446e4166b2f5e444a2b91ebc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4144eabef86647048c4d4ea9d897c444 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/278f82e100b74e53b57221b3a6c0f2d6 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"870fccd6b8a76844571d79581a1937d11ff13259","revisions":{"56bb96853cf699adc0c3db104682635dc2063fdb":{"kind":"REWORK","_number":1,"created":"2026-07-25 21:55:02.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/03/998703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/03/998703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/1"}}},"commit":{"parents":[{"commit":"ee6fd632b9e2548ee2dd969c7a4a3700146c9064","subject":"Merge \"Fix DictColumn JSON serialization error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/ee6fd632b9e2548ee2dd969c7a4a3700146c9064"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-07-25 21:40:56.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-07-25 21:51:56.000000000","tz":540},"subject":"Add availability_zone filter to share list (2.97)","message":"Add availability_zone filter to share list (2.97)\n\nManila added an availability_zone filter to the share list API in\nmicroversion 2.97. Wire it up in the client.\n\nRelated-Bug: #2154752\nChange-Id: I8dc604d83425f16aeb033d893336daea06915e52\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/56bb96853cf699adc0c3db104682635dc2063fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/56bb96853cf699adc0c3db104682635dc2063fdb"}]},"branch":"refs/heads/master"},"870fccd6b8a76844571d79581a1937d11ff13259":{"kind":"REWORK","_number":2,"created":"2026-08-27 00:40:45.000000000","uploader":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"ref":"refs/changes/03/998703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/03/998703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/03/998703/2"}}},"commit":{"parents":[{"commit":"39467871eb0ab72f64b491c5e44d96295a45fe84","subject":"Merge \"Fix resource lock project and user filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/39467871eb0ab72f64b491c5e44d96295a45fe84"}]}],"author":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-07-25 21:40:56.000000000","tz":540},"committer":{"name":"hongp","email":"inyong.hong@samsung.com","date":"2026-08-27 00:35:13.000000000","tz":540},"subject":"Add availability_zone filter to share list (2.97)","message":"Add availability_zone filter to share list (2.97)\n\nManila added an availability_zone filter to the share list API in\nmicroversion 2.97. Wire it up in the client.\n\nRelated-Bug: #2154752\nChange-Id: I8dc604d83425f16aeb033d893336daea06915e52\nSigned-off-by: hongp \u003cinyong.hong@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/870fccd6b8a76844571d79581a1937d11ff13259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/870fccd6b8a76844571d79581a1937d11ff13259"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
