)]}'
{"id":"openstack%2Fpython-masakariclient~1005979","triplet_id":"openstack%2Fpython-masakariclient~master~I29d982331b6e0f6c878f92288bb74f8ebba19ab7","project":"openstack/python-masakariclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-17 09:55:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I29d982331b6e0f6c878f92288bb74f8ebba19ab7","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2026-09-17 02:17:31.000000000","updated":"2026-09-17 10:09:55.000000000","submitted":"2026-09-17 10:08:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005979","meta_rev_id":"aaa33c18a5dc806e2bb61e31b3b5bf56e2d7353a","_number":1005979,"virtual_id_number":1005979,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 10:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-17 09:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-17 09:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 03:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4abbcb3cccbf7e4d2ea55485f9eacacceeb48e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-17 02:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c023e4718dff43cdd01cef4acb84e013484df107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 03:25:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a76ff4af5809427ba75a97079122f2a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4ec534dc663849d5b1a316dc5385a14a : FAILURE in 2m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/449b9169c51b4bf8b5af8f87ee203fc0 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f597bf1b56334768a2152acaaba10675 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fd22bee6d8c41c18856194bf9ba5c59 : FAILURE in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81a4a3e707a940b68660189d5ff08fe7 : FAILURE in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed5cd5f77df04b2b9eba7d099ced3272 : FAILURE in 5m 14s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e0a298fd26a04a26becab67f4969033b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3084c596391d4f4f876e18695b40d65a : FAILURE in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a42a987295c4ccf9ed1081629dc73d8 : SUCCESS in 2m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a3b761409fd4a7c91f8a4a4fdbdacca : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a860f63165de4256e6b798911e0ff3cb87154393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-17 06:34:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fa22cc0c78b992acbafe4c82bd5749026d02796d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 07:01:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38adc30178a2434381951d8019a8ecc4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/68f4dfd26f264ab7b00f874074659725 : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3bd15edd11c48ed92cfd3b7bf8c9fb5 : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e30445b72a4d3db6f470a867f4e30d : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86686190638d4ababab8cfc96fe7ede5 : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e75296b19124d3c9cb6423418ec1d92 : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dae45cc5afd84b14a01c4280c46cea77 : SUCCESS in 4m 58s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b09b8f9e02124c689ea71d3d9c2d49fc : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f26a51e3bdf49c08d355ffc69a34d75 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e1917231ac46f3865374b2fd258e52 : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da1bbf7eeda041bf9ece86404eae59f6 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1e9728b98ec31851705542c1b4aa696bbe812a95","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-17 09:55:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"39cab98303338c3936de3f01c1b1705fe8dad0ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:56:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f12aa2e92aa1dfa76157d879aeb93106c285d440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:08:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e3bb2a117949ffbb69a6c32485cb0f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35279d5606d2427fb7034f8dd8d365f1 : SUCCESS in 2m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b54f48b79b84fd798b0d6c801efb05d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e66699069d42bab79d3d08338189bb : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab3131f9bae145ea9f8d4abe544ae6fa : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b232ca269e2442bae8449dbced920d7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd09f51d1db49c3aff839a27d879944 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eae91b1a6be44ddbc12a2438b717742 : SUCCESS in 3m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4fb163468a6f4b64be26b3b0b2803f83 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5e5729ad4ba1962c405a0356ea75b5bd0b29b5cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:08:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"aaa33c18a5dc806e2bb61e31b3b5bf56e2d7353a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:09:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1816b1c423b94cb98f57e5e3a6b0b3a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b955d1ab483540ff9962978289981884 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5aedb389494bcc00277c775ab8929cdcf71e6692","revisions":{"df51f24a12cac140a6de38f7725a054a7f1a6528":{"kind":"REWORK","_number":1,"created":"2026-09-17 02:17:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/1005979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-masakariclient","ref":"refs/changes/79/1005979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/1"}}},"commit":{"parents":[{"commit":"217b38ece27627566872f2733e793e322243fdeb","subject":"Merge \"tox: Drop redundant passenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/217b38ece27627566872f2733e793e322243fdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-17 02:15:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-17 02:17:12.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I29d982331b6e0f6c878f92288bb74f8ebba19ab7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/df51f24a12cac140a6de38f7725a054a7f1a6528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/df51f24a12cac140a6de38f7725a054a7f1a6528"}]},"branch":"refs/heads/master"},"5aedb389494bcc00277c775ab8929cdcf71e6692":{"kind":"REWORK","_number":2,"created":"2026-09-17 06:34:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/1005979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-masakariclient","ref":"refs/changes/79/1005979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-masakariclient refs/changes/79/1005979/2"}}},"commit":{"parents":[{"commit":"217b38ece27627566872f2733e793e322243fdeb","subject":"Merge \"tox: Drop redundant passenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/217b38ece27627566872f2733e793e322243fdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-17 02:15:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-17 06:34:05.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I29d982331b6e0f6c878f92288bb74f8ebba19ab7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/5aedb389494bcc00277c775ab8929cdcf71e6692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/5aedb389494bcc00277c775ab8929cdcf71e6692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
