)]}'
{"id":"openstack%2Fpython-masakariclient~1006395","triplet_id":"openstack%2Fpython-masakariclient~master~I28360c108a61ff8a37bd5c805d75daeea42b6285","project":"openstack/python-masakariclient","branch":"master","hashtags":[],"change_id":"I28360c108a61ff8a37bd5c805d75daeea42b6285","subject":"tox: Drop basepython","status":"MERGED","created":"2026-09-20 03:38:57.000000000","updated":"2026-09-21 11:37:19.000000000","submitted":"2026-09-21 11:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006395","meta_rev_id":"8eb2f8798db687d9b3bcc54493fe6673827017d7","_number":1006395,"virtual_id_number":1006395,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 11:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-21 11:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-21 11:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-20 03:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2e11cf404c91a4daca3ec3f25f8faaab0c4998b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-20 03:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0e2b6d167bf262cbdb324d52e798e76b1c7ec37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 03:52:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24ae6b8d42964349bcf642332b5e857f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79b49293d6a4453b11a3f0e8be251f9 : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29ae1582e1f64750824c549b4ed4e058 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/134c995e928d40f697a05342523480b7 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb699957275a45aa96b0c684c6bf4021 : SUCCESS in 5m 55s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/58ea64fd591e4973b25ac9d6bfa46cd7 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e21c1c1d64e411a9ee7633d15a14cce : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5835fb740e6649128e6990004e57827c : SUCCESS in 2m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea411b512d514cd8818f915955c6f889 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c7e483287839de372f7779d0cc59f8804b67c584","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-21 11:26:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ace9fe9c07ba9d49a2ab708ad9d4d7972076e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:26:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3541da67f6681c1a3f5d658b60458b0bcbb5a4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:35:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba57cfb862a45248fb3a857a0075841\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d245b4d828942d1ac3aa2a5322d9e12 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e73b0c6c23a84440a1a3721138ccfba8 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5685080bb294aa382c17b0a811ecfcb : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49212008d9246aa8de668f2c741fa06 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1395a08b2207492bb3decafb9629dec3 : SUCCESS in 2m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae0f95a65e16460a811ec9cf831e40a0 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d61c7468539e75dfcad035257e4b1c8e11f03faa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:35:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8eb2f8798db687d9b3bcc54493fe6673827017d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:37:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c796dfdfa1c42b5920cb9281a5573e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f8675775c64779a4424435fd2534eb : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b724038b61b45b6bb0e7d4916e17d95 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b87da94588b8dd1335fee958f617cf1b675a0a8","revisions":{"5b87da94588b8dd1335fee958f617cf1b675a0a8":{"kind":"REWORK","_number":1,"created":"2026-09-20 03:38:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/1006395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-masakariclient","ref":"refs/changes/95/1006395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/95/1006395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/95/1006395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-masakariclient refs/changes/95/1006395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-masakariclient refs/changes/95/1006395/1"}}},"commit":{"parents":[{"commit":"217b38ece27627566872f2733e793e322243fdeb","subject":"Merge \"tox: Drop redundant passenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/217b38ece27627566872f2733e793e322243fdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-20 03:36:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-20 03:38:47.000000000","tz":540},"subject":"tox: Drop basepython","message":"tox: Drop basepython\n\nPython 2 reached its EOL long time ago and we no longer expect any\nuser may attempt to use it.\n\nChange-Id: I28360c108a61ff8a37bd5c805d75daeea42b6285\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/5b87da94588b8dd1335fee958f617cf1b675a0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-masakariclient/commit/5b87da94588b8dd1335fee958f617cf1b675a0a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
