)]}'
{"id":"openstack%2Fpython-mistralclient~1005449","triplet_id":"openstack%2Fpython-mistralclient~master~Iab70c29a497dca553ced85f0e6bf9bf3f0a7593c","project":"openstack/python-mistralclient","branch":"master","hashtags":[],"change_id":"Iab70c29a497dca553ced85f0e6bf9bf3f0a7593c","subject":"Remove unused functional test script","status":"NEW","created":"2026-09-13 19:01:53.000000000","updated":"2026-09-13 19:27:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15a6aa899eb364bbd2e4e61467cae3e55a385e5b","_number":1005449,"virtual_id_number":1005449,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-13 19:27:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 19:27:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95c85492fb6c2ead4605ba7ebe8e206c68d1af4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 19:01:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15a6aa899eb364bbd2e4e61467cae3e55a385e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 19:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa808424bcc4a4995dab61259644d78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a512d9e926d42cb9f84437a186edec7 : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a69e5f407ef34972b30b09dc25ca039a : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4eb6a3f2509d4d288da1189ec78c7e0d : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/146c8c29bdd64bfa912b1fac04733223 : SUCCESS in 3m 45s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/480aeea76586470e880a15fd2c7cf887 : SUCCESS in 4m 29s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/639a73f359e74d3da7be88365da603d6 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d20b162b38428881760e078b919939 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e714c9827924d76bb0686fb5304e688 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73d5e13aba34407e8f9cfe87f3cfdf7a : SUCCESS in 2m 45s (non-voting)\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/18f40014ce1e49178c9e4345ad89a70f : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5285806b8cd6f22f9e756198d2f00c52b99d7a01","revisions":{"5285806b8cd6f22f9e756198d2f00c52b99d7a01":{"kind":"REWORK","_number":1,"created":"2026-09-13 19:01:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/1005449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-mistralclient","ref":"refs/changes/49/1005449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/49/1005449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/49/1005449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/49/1005449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-mistralclient refs/changes/49/1005449/1"}}},"commit":{"parents":[{"commit":"575367c2b51fc048598c6bf9434dab693173743a","subject":"Merge \"tox: Drop redundant passenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/575367c2b51fc048598c6bf9434dab693173743a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:59:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 19:01:31.000000000","tz":540},"subject":"Remove unused functional test script","message":"Remove unused functional test script\n\nThe script isn\u0027t used in CI, and does no longer work due to its\nremaining usage of nose.\n\nChange-Id: Iab70c29a497dca553ced85f0e6bf9bf3f0a7593c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/5285806b8cd6f22f9e756198d2f00c52b99d7a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/5285806b8cd6f22f9e756198d2f00c52b99d7a01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
