)]}'
{"id":"openstack%2Fpython-mistralclient~1005473","triplet_id":"openstack%2Fpython-mistralclient~master~I1a4f1fcf9864abd5fe7154cfd474361f17e5df4a","project":"openstack/python-mistralclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 13:48:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a4f1fcf9864abd5fe7154cfd474361f17e5df4a","subject":"Add minimum pyproject.toml","status":"MERGED","created":"2026-09-14 02:51:43.000000000","updated":"2026-09-14 13:49:06.000000000","submitted":"2026-09-14 13:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005473","meta_rev_id":"cf38e8568004ebe193b0509e74fc8262744f9d8e","_number":1005473,"virtual_id_number":1005473,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 13:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-14 12:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-14 12:55:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 03:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 12:55:58.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"d2af9f3e2c849c76240ca3d1964a00c7ba8437cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 02:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"105de6ba5998fe60c974df9a723e0b2097d50b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 03:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a27ecb5303d744088155f5dce375b143\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba543a47e79744a7a66a640b6398a973 : SUCCESS in 3m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da16a80bdf7f4e14bb27b85cea6552a4 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0223bf5aa53460e83d844a84cda81e7 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9eb425a15eb485b83e1adfe994dc6b3 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37c17dcd3cd14ac4b144e36021f69f89 : SUCCESS in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc596ae5170740fe9a53f10318acf80c : SUCCESS in 6m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ba885bbc3d944523a958f930b04ad2f3 : SUCCESS in 5m 11s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4a03d6a38d904b2cbfd4e390118c75b4 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6e33d176dc45808b2fe4f05b8e0f26 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c851ea41f84b6ba7b95102b0d3d22c : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dc2618156684e469636e995a7c539e6 : SUCCESS in 4m 02s (non-voting)\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/9a0eef0174a749cb9282a600d404312e : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"77e306bd99e4f244f828f2afe0b9c02d0a086327","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-14 12:55:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e69014b0913a70e39e468deb7873c8f0c2b0ed3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:56:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4881376d16d9b9794ae4afaa6fa74abd0b7f0b60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:48:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7660263dfeca46858e76d49bee307876\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bfb5e2e1d9694387afe2c67459eb973d : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e157a0e5c0d54115a137faf7b2869729 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319f713520b24fb6bcdaed48d8810e54 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6dfe0d3835a6421a9fcba7ba091f38b6 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70bf4539d853474880312aecdb521682 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20e62f7309754cc6af86fc868f500e71 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e416b5156947f6bc29e7baf7f9d229 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6b9dd5a32f14eccb0c0477d945133f2 : SUCCESS in 3m 54s\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/70ccb6809c864c54a4c8361a80e7bdf3 : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"822c3dfd6c183d97006c065b13528560de97699a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:48:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf38e8568004ebe193b0509e74fc8262744f9d8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:49:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539d36d4113d452daba0c602d71d3fda\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef40ed659a394214944f9508b4ef32d2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87fb83f8d03a41c197edf0e309b918603356046f","revisions":{"87fb83f8d03a41c197edf0e309b918603356046f":{"kind":"REWORK","_number":1,"created":"2026-09-14 02:51:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/1005473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-mistralclient","ref":"refs/changes/73/1005473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/73/1005473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/73/1005473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/73/1005473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-mistralclient refs/changes/73/1005473/1"}}},"commit":{"parents":[{"commit":"575367c2b51fc048598c6bf9434dab693173743a","subject":"Merge \"tox: Drop redundant passenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/575367c2b51fc048598c6bf9434dab693173743a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 02:51:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 02:51:31.000000000","tz":540},"subject":"Add minimum pyproject.toml","message":"Add minimum pyproject.toml\n\nThe file is now required by the requirements-check job.\n\nChange-Id: I1a4f1fcf9864abd5fe7154cfd474361f17e5df4a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/87fb83f8d03a41c197edf0e309b918603356046f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/87fb83f8d03a41c197edf0e309b918603356046f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
