)]}'
{"id":"openstack%2Fpython-mistralclient~976877","triplet_id":"openstack%2Fpython-mistralclient~master~I2f3cb7a2022d2f39032b33456f2e5950685ee3d6","project":"openstack/python-mistralclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-11 08:59:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f3cb7a2022d2f39032b33456f2e5950685ee3d6","subject":"tox: Drop redundant injection of VIRTUAL_ENV variable","status":"MERGED","created":"2026-02-15 14:16:57.000000000","updated":"2026-07-11 08:59:18.000000000","submitted":"2026-07-11 08:59:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976877","meta_rev_id":"6dca8e4ef6d6b71161d7d972c1038edb04b26780","_number":976877,"virtual_id_number":976877,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-11 08:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-11 08:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-11 08:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-15 14:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 08:00:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"ce5151e6c2d3de627fdf44edf325b2dd67c0ff00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-15 14:16:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f88b54fbcd000b8eaf0dee6308af04c394d2437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 14:59:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df3861ca65b64affbafcb3fb38bcca8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291fc257eda94c04bbe36101498d6993 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea372ea6806442969fb15a24ccfa4c19 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0caa0c1a857e4b5d9f10c3d3d9f4f6dd : SUCCESS in 3m 36s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/457a62b4eb5a4098bc11f29aef4973e7 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe0c17a74ad486aa7e4b37b6ec4c1d4 : SUCCESS in 1m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e8b777207714ebf82712fe1617ed66e : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac89b58b4d141cfac98f0c7ae166b49 : SUCCESS in 4m 38s (non-voting)\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/e22f515a173b4f6988aa34e46f1ef849 : FAILURE in 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2b1502ccfc79142933c7d78e5e9a7f672d0e9a2c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-16 07:16:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f73e60490926e1bfffd6251e2ab809b375b9a652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 07:49:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8d507c53cce40be85f6e0a88916349c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ce0a749b1040d994a1609520392075 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd2f9e39115a4ef8990a4958535cdf3c : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/684688b51f4841bb9bd36342678d792a : SUCCESS in 3m 14s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4705490a93804ad497d8f497d64abbdf : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ffcb55c6a864a38acd79fc318db382e : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ad549b45b354f60a69eb8654ffc32bc : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44eb9c47da534391a2692ba984727437 : SUCCESS in 3m 08s (non-voting)\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/98608d1df99c4fa1ae9e531701d88358 : FAILURE in 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"263c772bea6fca359fef88615ed52244488c340c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-07-10 21:33:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"036a88f9f85163ed13697b30bdcc4b9a188e04a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 22:25:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dabb526c23f409d983b6e8bf0cbe213\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0a960b4ac14109a864e60d7b8053f7 : SUCCESS in 2m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b330e60dbfae4ca19884328a3eaeea98 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b96fffd6369f403bb3a0328e40fbccd8 : SUCCESS in 2m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8702fcdf11f94fcba6a046584eb95cf2 : SUCCESS in 3m 25s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a35966d6a75b4aaeb8111f944a951701 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9e5ed395d54d5f8248ec3672d2ef16 : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f5f4da98937404e9e8a6f7b96d01f78 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45b6ce8a044247fda319e0d67b45a2f9 : SUCCESS in 3m 39s (non-voting)\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/5b67051032cc4d2cb3aaf4bab37d1771 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1163aa57a3e28820f010a2b91a22ca0e3d314168","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-07-11 08:00:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e84d47b4b17af4c646fe5d637f9ae61579b52b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:02:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9f328830506f2b71df0c5807b5362b0580a103a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:59:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b4196e6cb84d72ae97a82d925d11e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad3408ca0ab430bb48fb532bc44d9d0 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57c5016ae27c40f383b29267088affe2 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aca6db80567544f5b8407cab47203e7a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5397b6611842318a7a95d69a9ba313 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2244b60fb51e4ca194f6fa4d30118a0c : SUCCESS in 3m 59s\n- python-mistralclient-functional-devstack https://zuul.opendev.org/t/openstack/build/94111528d48744cc8604676dc8563ca5 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":2},{"id":"6dca8e4ef6d6b71161d7d972c1038edb04b26780","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 08:59:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d2a961943459a66d50cea2725796e6b6f3370b8","revisions":{"ec7c601bfb25b91b00395ef2dd0e6c14bf075c77":{"kind":"REWORK","_number":1,"created":"2026-02-15 14:16:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/976877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-mistralclient","ref":"refs/changes/77/976877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/1"}}},"commit":{"parents":[{"commit":"1cb8e3ac630e2aa1bbff4d9d2c16842c8fec9edf","subject":"Drop support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/1cb8e3ac630e2aa1bbff4d9d2c16842c8fec9edf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 14:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 14:16:48.000000000","tz":540},"subject":"tox: Drop redundant injection of VIRTUAL_ENV variable","message":"tox: Drop redundant injection of VIRTUAL_ENV variable\n\nThe environment variable is set automatically.\n\ncf. https://github.com/tox-dev/tox/commit/be5373bab\n\nChange-Id: I2f3cb7a2022d2f39032b33456f2e5950685ee3d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/ec7c601bfb25b91b00395ef2dd0e6c14bf075c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/ec7c601bfb25b91b00395ef2dd0e6c14bf075c77"}]},"branch":"refs/heads/master"},"2d2a961943459a66d50cea2725796e6b6f3370b8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-10 21:33:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/77/976877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-mistralclient","ref":"refs/changes/77/976877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-mistralclient refs/changes/77/976877/2"}}},"commit":{"parents":[{"commit":"5b82d7e2d40bd58a4fa5abd4d211671acf696c98","subject":"Make CI jobs voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/5b82d7e2d40bd58a4fa5abd4d211671acf696c98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 14:16:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-10 21:33:01.000000000","tz":0},"subject":"tox: Drop redundant injection of VIRTUAL_ENV variable","message":"tox: Drop redundant injection of VIRTUAL_ENV variable\n\nThe environment variable is set automatically.\n\ncf. https://github.com/tox-dev/tox/commit/be5373bab\n\nChange-Id: I2f3cb7a2022d2f39032b33456f2e5950685ee3d6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/2d2a961943459a66d50cea2725796e6b6f3370b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-mistralclient/commit/2d2a961943459a66d50cea2725796e6b6f3370b8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1cb8e3ac630e2aa1bbff4d9d2c16842c8fec9edf","ours":"ec7c601bfb25b91b00395ef2dd0e6c14bf075c77","theirs":"5b82d7e2d40bd58a4fa5abd4d211671acf696c98","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
