)]}'
{"id":"openstack%2Fpython-monascaclient~571716","triplet_id":"openstack%2Fpython-monascaclient~master~Ic9317d0d2e71ffcc5012184510b554453a7e322f","project":"openstack/python-monascaclient","branch":"master","hashtags":[],"change_id":"Ic9317d0d2e71ffcc5012184510b554453a7e322f","subject":"Switch to using stestr","status":"MERGED","created":"2018-06-01 12:59:33.000000000","updated":"2018-06-05 08:33:46.000000000","submitted":"2018-06-05 08:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"571716-1528187626388-386d9f20","meta_rev_id":"baaa40bdede96143cdc0bc36edfeb56159848d41","_number":571716,"virtual_id_number":571716,"owner":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2018-06-05 08:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2018-06-04 10:37:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-06-04 08:48:04.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-05 08:25:33.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-04 10:37:10.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2018-06-05 08:25:33.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2018-06-05 08:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f81bc4c8310189424ff4b4db932f84287d9706c4","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-06-01 12:59:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"050c32b894a9af99503dfb06470fa7089a303e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-01 13:08:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/16/571716/1/check/openstack-tox-pep8/0c4bcaa/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/16/571716/1/check/openstack-tox-py27/91c495f/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/16/571716/1/check/openstack-tox-py35/0016693/ : SUCCESS in 4m 05s\n- requirements-check http://logs.openstack.org/16/571716/1/check/requirements-check/20945e0/ : FAILURE in 3m 03s\n- build-openstack-releasenotes http://logs.openstack.org/16/571716/1/check/build-openstack-releasenotes/7e24af8/html/ : SUCCESS in 2m 52s\n- openstack-tox-cover http://logs.openstack.org/16/571716/1/check/openstack-tox-cover/b0a3930/cover/ : SUCCESS in 4m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/571716/1/check/build-openstack-sphinx-docs/87744c0/html/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/571716/1/check/openstack-tox-lower-constraints/15e7db4/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d8860ad96c3112b4aec8c37bae9ca5c5bc2dc76e","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-06-03 01:04:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58d2d83303b97eaf9c6c06752a1d2a55f2ae6e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-03 01:13:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/16/571716/2/check/openstack-tox-pep8/952d7b8/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/16/571716/2/check/openstack-tox-py27/e6d553d/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/16/571716/2/check/openstack-tox-py35/4eaedbc/ : SUCCESS in 3m 42s\n- requirements-check http://logs.openstack.org/16/571716/2/check/requirements-check/dd30a10/ : FAILURE in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/16/571716/2/check/build-openstack-releasenotes/aef05c4/html/ : SUCCESS in 3m 06s\n- openstack-tox-cover http://logs.openstack.org/16/571716/2/check/openstack-tox-cover/5c90903/cover/ : SUCCESS in 3m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/571716/2/check/build-openstack-sphinx-docs/35584dc/html/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/571716/2/check/openstack-tox-lower-constraints/4fda967/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a41adbbe4e35d2821975002c32f252fbdfe507ee","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-06-03 01:19:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e855051e2e8434a98e8adceb9c2f4f415ea924f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-03 01:27:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/16/571716/3/check/openstack-tox-pep8/8e4bc6b/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/16/571716/3/check/openstack-tox-py27/53d0d6a/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/16/571716/3/check/openstack-tox-py35/2bf702c/ : SUCCESS in 4m 08s\n- requirements-check http://logs.openstack.org/16/571716/3/check/requirements-check/27c42aa/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/16/571716/3/check/build-openstack-releasenotes/163506e/html/ : SUCCESS in 2m 56s\n- openstack-tox-cover http://logs.openstack.org/16/571716/3/check/openstack-tox-cover/17d883b/cover/ : SUCCESS in 3m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/571716/3/check/build-openstack-sphinx-docs/cfc8c10/html/ : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/571716/3/check/openstack-tox-lower-constraints/38b64a4/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f839b1a5393f022bc4918cfc1529533f31919a4b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-04 08:48:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84c10913dd4b1f6151a4ca21f93a2ff26aabe149","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-06-04 10:37:10.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks like there\u0027s an issue with one of the unit tests, but\u0027s unrelated to this change. It doesn\u0027t fail, but generates this:\n\n```\n{0} monascaclient.tests.test_shell.TestMonascaShell.test_should_specify_monasca_args [0.000649s] ... ok\n   ERROR [run] \u0027Mock\u0027 object is not iterable\nTraceback (most recent call last):\n  File \"/home/doug/git/python-monascaclient/.tox/py35/lib/python3.5/site-packages/cliff/app.py\", line 246, in run\n    self.options, remainder \u003d self.parser.parse_known_args(argv)\nTypeError: \u0027Mock\u0027 object is not iterable\n   ERROR [osc_lib.shell] Traceback (most recent call last):\n  File \"/home/doug/git/python-monascaclient/.tox/py35/lib/python3.5/site-packages/osc_lib/shell.py\", line 134, in run\n    ret_val \u003d super(OpenStackShell, self).run(argv)\n  File \"/home/doug/git/python-monascaclient/.tox/py35/lib/python3.5/site-packages/cliff/app.py\", line 246, in run\n    self.options, remainder \u003d self.parser.parse_known_args(argv)\nTypeError: \u0027Mock\u0027 object is not iterable\n```","accounts_in_message":[],"_revision_number":3},{"id":"408bffcf12238ef3852121c5e34bd1a11000a385","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-05 08:25:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2219d73a0db60e100d361f42ef5ceb56821796b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 08:25:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70560590dbabe247d896a11165d7f6261d4b265a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 08:33:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/16/571716/3/gate/openstack-tox-pep8/f0e6f27/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/16/571716/3/gate/openstack-tox-py27/f024efe/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/16/571716/3/gate/openstack-tox-py35/4c4d306/ : SUCCESS in 4m 35s\n- requirements-check http://logs.openstack.org/16/571716/3/gate/requirements-check/2c8cabc/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/16/571716/3/gate/build-openstack-releasenotes/8e25523/html/ : SUCCESS in 3m 36s\n- openstack-tox-cover http://logs.openstack.org/16/571716/3/gate/openstack-tox-cover/e1821ae/cover/ : SUCCESS in 4m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/571716/3/gate/build-openstack-sphinx-docs/7774ba1/html/ : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/571716/3/gate/openstack-tox-lower-constraints/f888dbe/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"baaa40bdede96143cdc0bc36edfeb56159848d41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-05 08:33:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ff3520baa0273da8e45fe45953fa7ad4af87089c","revisions":{"997530b75996ea6315288e742df5db2f0de2025b":{"kind":"REWORK","_number":1,"created":"2018-06-01 12:59:33.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/16/571716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-monascaclient","ref":"refs/changes/16/571716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/1"}}},"commit":{"parents":[{"commit":"749362a4cddb5a95fbb0f64fc744f80da5842fa8","subject":"Merge \"Update links in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/749362a4cddb5a95fbb0f64fc744f80da5842fa8"}]}],"author":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-01 12:56:46.000000000","tz":-240},"committer":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-01 12:59:32.000000000","tz":-240},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is\nmaintained project to which all Openstack projects\nshould migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: Ic9317d0d2e71ffcc5012184510b554453a7e322f\nSigned-off-by: Charles Short \u003czulcss@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/997530b75996ea6315288e742df5db2f0de2025b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/997530b75996ea6315288e742df5db2f0de2025b"}]},"branch":"refs/heads/master"},"e1b5fec6867283da226c1807ad566b99b34a32c3":{"kind":"REWORK","_number":2,"created":"2018-06-03 01:04:22.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/16/571716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-monascaclient","ref":"refs/changes/16/571716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/2"}}},"commit":{"parents":[{"commit":"749362a4cddb5a95fbb0f64fc744f80da5842fa8","subject":"Merge \"Update links in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/749362a4cddb5a95fbb0f64fc744f80da5842fa8"}]}],"author":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-01 12:56:46.000000000","tz":-240},"committer":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-03 01:04:13.000000000","tz":-240},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is\nmaintained project to which all Openstack projects\nshould migrate.\nLet\u0027s switch it then.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: Ic9317d0d2e71ffcc5012184510b554453a7e322f\nSigned-off-by: Charles Short \u003czulcss@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/e1b5fec6867283da226c1807ad566b99b34a32c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/e1b5fec6867283da226c1807ad566b99b34a32c3"}]},"branch":"refs/heads/master"},"ff3520baa0273da8e45fe45953fa7ad4af87089c":{"kind":"REWORK","_number":3,"created":"2018-06-03 01:19:07.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/16/571716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-monascaclient","ref":"refs/changes/16/571716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-monascaclient refs/changes/16/571716/3"}}},"commit":{"parents":[{"commit":"749362a4cddb5a95fbb0f64fc744f80da5842fa8","subject":"Merge \"Update links in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/749362a4cddb5a95fbb0f64fc744f80da5842fa8"}]}],"author":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-01 12:56:46.000000000","tz":-240},"committer":{"name":"Charles Short","email":"zulcss@gmail.com","date":"2018-06-03 01:18:55.000000000","tz":-240},"subject":"Switch to using stestr","message":"Switch to using stestr\n\nAccording to Openstack summit session [1] stestr is\nmaintained project to which all Openstack projects\nshould migrate.\nLet\u0027s switch it then.\n\nAlso clean up requirements.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: Ic9317d0d2e71ffcc5012184510b554453a7e322f\nSigned-off-by: Charles Short \u003czulcss@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/ff3520baa0273da8e45fe45953fa7ad4af87089c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-monascaclient/commit/ff3520baa0273da8e45fe45953fa7ad4af87089c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
