)]}'
{"id":"openstack%2Fpython-muranoclient~664492","triplet_id":"openstack%2Fpython-muranoclient~master~Ia15cc48ce020ff936185d6a94312e68c3c1603b3","project":"openstack/python-muranoclient","branch":"master","hashtags":[],"change_id":"Ia15cc48ce020ff936185d6a94312e68c3c1603b3","subject":"Bump the sphinx to 1.6.2 for lower-constraints.txt","status":"MERGED","created":"2019-06-11 04:10:17.000000000","updated":"2019-06-14 06:59:49.000000000","submitted":"2019-06-14 06:59:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664492-1560495589440-b1f8f8f4","meta_rev_id":"1725ca5fe02115752f4ae90438c88cc3b5fac247","_number":664492,"virtual_id_number":664492,"owner":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-14 06:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-06-14 06:21:34.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-06-14 06:21:34.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 06:21:34.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-06-14 06:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a00cf3cc5c71a2accaa034447bc89bd813a7814","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-11 04:10:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf7b103609cc418b32584d83631053cb179863ec","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-11 04:14:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"255b18cd6c5accf459bf5a1b794d86ee5968f1ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:02:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/92/664492/2/check/openstack-tox-cover/63a15ad/cover/ : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/664492/2/check/openstack-tox-lower-constraints/f812216/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/92/664492/2/check/openstack-tox-pep8/fa2c052/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/92/664492/2/check/openstack-tox-py27/c78af1f/ : SUCCESS in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/92/664492/2/check/openstack-tox-py36/fa4f651/ : SUCCESS in 5m 56s\n- openstack-tox-py37 http://logs.openstack.org/92/664492/2/check/openstack-tox-py37/399de11/ : SUCCESS in 4m 50s\n- requirements-check http://logs.openstack.org/92/664492/2/check/requirements-check/718acb6/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/92/664492/2/check/openstack-tox-docs/78ee311/html/ : SUCCESS in 5m 07s\n- openstackclient-check-plugins http://logs.openstack.org/92/664492/2/check/openstackclient-check-plugins/118b2f3/ : SUCCESS in 6m 45s (non-voting)\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/92/664492/2/check/muranoclient-functional-test-mysql-backend/900aae5/ : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ce6ad139f0670e06043b497ec2c75832a0ff10de","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-14 06:21:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"763bd50c8406035888011d601b86e5ef4624c284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:21:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b04ece982c2d927a1e861cd4b570d4cb72590cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:59:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/92/664492/2/gate/openstack-tox-lower-constraints/b056ebb/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/92/664492/2/gate/openstack-tox-pep8/5bbb51f/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/92/664492/2/gate/openstack-tox-py27/cbd94d7/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/92/664492/2/gate/openstack-tox-py36/935b811/ : SUCCESS in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/92/664492/2/gate/openstack-tox-py37/fe994c9/ : SUCCESS in 5m 23s\n- requirements-check http://logs.openstack.org/92/664492/2/gate/requirements-check/846f5ad/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/92/664492/2/gate/openstack-tox-docs/b6e45e3/html/ : SUCCESS in 4m 35s\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/92/664492/2/gate/muranoclient-functional-test-mysql-backend/6e4f45f/ : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1725ca5fe02115752f4ae90438c88cc3b5fac247","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:59:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"167545e89ea7b7c69f56e4dbd256b80a04fd7e54","revisions":{"7aa40a40af32e65fbd63b3c59aaf5e141a4db579":{"kind":"REWORK","_number":1,"created":"2019-06-11 04:10:17.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/92/664492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-muranoclient","ref":"refs/changes/92/664492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/1"}}},"commit":{"parents":[{"commit":"9a6a86ed72dfcffbf9221ab2990457b9f9ea762b","subject":"Merge \"Add irrelevant-files for muranoclient-functional-test-mysql-backend job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/9a6a86ed72dfcffbf9221ab2990457b9f9ea762b"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 04:08:47.000000000","tz":0},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 04:08:47.000000000","tz":0},"subject":"Bump the sphinx to 1.6.2 for lower-constraints.txt","message":"Bump the sphinx to 1.6.2 for lower-constraints.txt\n\n1. refer to https://github.com/openstack/python-muranoclient/blob/master/doc/requirements.txt#L5\n\nChange-Id: Ia15cc48ce020ff936185d6a94312e68c3c1603b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/7aa40a40af32e65fbd63b3c59aaf5e141a4db579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/7aa40a40af32e65fbd63b3c59aaf5e141a4db579"}]},"branch":"refs/heads/master"},"167545e89ea7b7c69f56e4dbd256b80a04fd7e54":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-11 04:14:10.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/92/664492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-muranoclient","ref":"refs/changes/92/664492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-muranoclient refs/changes/92/664492/2"}}},"commit":{"parents":[{"commit":"9a6a86ed72dfcffbf9221ab2990457b9f9ea762b","subject":"Merge \"Add irrelevant-files for muranoclient-functional-test-mysql-backend job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/9a6a86ed72dfcffbf9221ab2990457b9f9ea762b"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 04:08:47.000000000","tz":0},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2019-06-11 04:13:05.000000000","tz":0},"subject":"Bump the sphinx to 1.6.2 for lower-constraints.txt","message":"Bump the sphinx to 1.6.2 for lower-constraints.txt\n\n1. Package \u0027sphinx\u0027 is constrained to 1.6.5 which does not match the\nminimum version specifier 1.6.2 in doc/requirements.txt[0]\n\n[0]: refer to https://github.com/openstack/python-muranoclient/blob/master/doc/requirements.txt#L5\n\nChange-Id: Ia15cc48ce020ff936185d6a94312e68c3c1603b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/167545e89ea7b7c69f56e4dbd256b80a04fd7e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/167545e89ea7b7c69f56e4dbd256b80a04fd7e54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
