)]}'
{"id":"openstack%2Fpython-muranoclient~665891","triplet_id":"openstack%2Fpython-muranoclient~master~I8be79e80d694715b5192944c140b21f92d4b72d9","project":"openstack/python-muranoclient","branch":"master","hashtags":[],"change_id":"I8be79e80d694715b5192944c140b21f92d4b72d9","subject":"Update contraints url","status":"MERGED","created":"2019-06-18 07:07:35.000000000","updated":"2019-06-18 09:46:55.000000000","submitted":"2019-06-18 09:46:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665891-1560851215214-eb0806c3","meta_rev_id":"0e6c5b4f81d7fe905301334cd93cb98038544717","_number":665891,"virtual_id_number":665891,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-18 09:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-06-18 08:47:55.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-06-18 08:47:55.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 09:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bece64c15a8a5ff0c78df1dbe970d3aad78abcfc","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 07:07:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d185ab8c1041cd4a562afef97e67a79e6b79c007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:55:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/665891/1/check/openstack-tox-cover/1879c14/cover/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665891/1/check/openstack-tox-lower-constraints/824c914/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/91/665891/1/check/openstack-tox-pep8/73d79e3/ : SUCCESS in 6m 56s\n- openstack-tox-py27 http://logs.openstack.org/91/665891/1/check/openstack-tox-py27/f73e3e8/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/91/665891/1/check/openstack-tox-py36/f068742/ : SUCCESS in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/91/665891/1/check/openstack-tox-py37/3863f1b/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/91/665891/1/check/build-openstack-releasenotes/12a096e/html/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/91/665891/1/check/openstack-tox-docs/a12bff1/html/ : SUCCESS in 6m 20s\n- openstackclient-check-plugins http://logs.openstack.org/91/665891/1/check/openstackclient-check-plugins/74eaf5f/ : SUCCESS in 5m 10s (non-voting)\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/91/665891/1/check/muranoclient-functional-test-mysql-backend/b3647ae/ : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1723bdc7d5fed6dc2a7a121a6ce174541f288d0a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-18 08:47:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1e4404b2a4ed2b52bc77d501c545ade322b9434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:48:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c76cbc47281ee8377e9ee7312e4dcb69bb03cab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:46:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665891/1/gate/openstack-tox-lower-constraints/db351ef/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/91/665891/1/gate/openstack-tox-pep8/7865163/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/91/665891/1/gate/openstack-tox-py27/299830c/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/91/665891/1/gate/openstack-tox-py36/9283273/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/91/665891/1/gate/openstack-tox-py37/7eba90b/ : SUCCESS in 5m 05s\n- build-openstack-releasenotes http://logs.openstack.org/91/665891/1/gate/build-openstack-releasenotes/1b13e54/html/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/91/665891/1/gate/openstack-tox-docs/f186400/html/ : SUCCESS in 4m 37s\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/91/665891/1/gate/muranoclient-functional-test-mysql-backend/a014424/ : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0e6c5b4f81d7fe905301334cd93cb98038544717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:46:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8","revisions":{"89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8":{"kind":"REWORK","_number":1,"created":"2019-06-18 07:07:35.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/91/665891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-muranoclient","ref":"refs/changes/91/665891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/91/665891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/91/665891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/91/665891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-muranoclient refs/changes/91/665891/1"}}},"commit":{"parents":[{"commit":"167545e89ea7b7c69f56e4dbd256b80a04fd7e54","subject":"Bump the sphinx to 1.6.2 for lower-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/167545e89ea7b7c69f56e4dbd256b80a04fd7e54"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:59:28.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-06-18 06:59:28.000000000","tz":480},"subject":"Update contraints url","message":"Update contraints url\n\nChange-Id: I8be79e80d694715b5192944c140b21f92d4b72d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
