)]}'
{"id":"openstack%2Fpython-muranoclient~668838","triplet_id":"openstack%2Fpython-muranoclient~master~I7234a0d87b66095a2acabbb0f17cbc0438de0ee0","project":"openstack/python-muranoclient","branch":"master","hashtags":[],"change_id":"I7234a0d87b66095a2acabbb0f17cbc0438de0ee0","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-03 09:19:15.000000000","updated":"2019-07-04 03:28:46.000000000","submitted":"2019-07-04 03:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668838-1562210926538-5870c8eb","meta_rev_id":"a2d0a79ee782c55d8ffb8af3b95986105593ea14","_number":668838,"virtual_id_number":668838,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-07-04 03:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-07-04 02:36:02.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-07-04 02:36:02.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 03:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"802256bcc37136cf712d72ddac525346c72f963b","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-03 09:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56c0a3cc6f27db0bf870432b91fb55f7feefbe1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/38/668838/1/check/test-release-openstack/ca6d2b6/ : SUCCESS in 1m 53s\n- openstack-tox-cover http://logs.openstack.org/38/668838/1/check/openstack-tox-cover/fdd71d9/cover/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668838/1/check/openstack-tox-lower-constraints/51ad444/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/38/668838/1/check/openstack-tox-pep8/a6a0457/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/38/668838/1/check/openstack-tox-py27/4a63383/ : FAILURE in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/38/668838/1/check/openstack-tox-py36/65f5fa5/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/38/668838/1/check/openstack-tox-py37/39c987a/ : SUCCESS in 3m 57s\n- build-openstack-releasenotes http://logs.openstack.org/38/668838/1/check/build-openstack-releasenotes/7782fad/html/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/38/668838/1/check/openstack-tox-docs/e2c3eab/html/ : SUCCESS in 2m 46s\n- openstackclient-check-plugins http://logs.openstack.org/38/668838/1/check/openstackclient-check-plugins/7f0197a/ : SUCCESS in 3m 53s (non-voting)\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/38/668838/1/check/muranoclient-functional-test-mysql-backend/6417802/ : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6a3aaae419802af9cac74e44959678dc88e6da75","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 00:52:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f4d0537020bff9bdee9572b51de0b8f9f2e44e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:30:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/38/668838/1/check/test-release-openstack/1a52ab0/ : SUCCESS in 2m 06s\n- openstack-tox-cover http://logs.openstack.org/38/668838/1/check/openstack-tox-cover/a799077/cover/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668838/1/check/openstack-tox-lower-constraints/84fa2be/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/38/668838/1/check/openstack-tox-pep8/fa43ceb/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/38/668838/1/check/openstack-tox-py27/5e8d9c7/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/38/668838/1/check/openstack-tox-py36/b556398/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/38/668838/1/check/openstack-tox-py37/a0ee256/ : SUCCESS in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/38/668838/1/check/build-openstack-releasenotes/51fac4b/html/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/38/668838/1/check/openstack-tox-docs/d55e667/html/ : SUCCESS in 2m 41s\n- openstackclient-check-plugins http://logs.openstack.org/38/668838/1/check/openstackclient-check-plugins/a931da1/ : SUCCESS in 4m 01s (non-voting)\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/38/668838/1/check/muranoclient-functional-test-mysql-backend/b886161/ : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8c7f274acef4e600e55e7fced197e7ced91feba4","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 02:36:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f6ab5a374cb943e066a8047c9c5d90177f74cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 02:36:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9df75eb95c38cee1dcfa2c40cb9c109ec24ac73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 03:28:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/38/668838/1/gate/test-release-openstack/e365236/ : SUCCESS in 2m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668838/1/gate/openstack-tox-lower-constraints/2dd3420/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/38/668838/1/gate/openstack-tox-pep8/0432b6d/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/38/668838/1/gate/openstack-tox-py27/f433330/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/38/668838/1/gate/openstack-tox-py36/2b24d44/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/38/668838/1/gate/openstack-tox-py37/7aa63be/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/38/668838/1/gate/build-openstack-releasenotes/ee5ba1a/html/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/38/668838/1/gate/openstack-tox-docs/0a53870/html/ : SUCCESS in 2m 38s\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/38/668838/1/gate/muranoclient-functional-test-mysql-backend/36ef2ce/ : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a2d0a79ee782c55d8ffb8af3b95986105593ea14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 03:28:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4db579c4be4a94c74a958857b33fa1d5d972f6c","revisions":{"e4db579c4be4a94c74a958857b33fa1d5d972f6c":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:19:15.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/38/668838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-muranoclient","ref":"refs/changes/38/668838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/38/668838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/38/668838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/38/668838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-muranoclient refs/changes/38/668838/1"}}},"commit":{"parents":[{"commit":"89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8","subject":"Update contraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/89122a7f3bc8ce10fbd0cb06358c2898ca3a2bb8"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-03 09:18:31.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-07-03 09:18:31.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I7234a0d87b66095a2acabbb0f17cbc0438de0ee0\nStory: #2005924\nTask: #34214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/e4db579c4be4a94c74a958857b33fa1d5d972f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/e4db579c4be4a94c74a958857b33fa1d5d972f6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
