)]}'
{"id":"openstack%2Fpython-muranoclient~669096","triplet_id":"openstack%2Fpython-muranoclient~master~If2ca7919c0a4a488b2719e4063a2c700f23d3e59","project":"openstack/python-muranoclient","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"If2ca7919c0a4a488b2719e4063a2c700f23d3e59","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-07-04 08:21:13.000000000","updated":"2019-07-04 12:26:44.000000000","submitted":"2019-07-04 12:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669096-1562243204911-577aa814","meta_rev_id":"fa1066a0568eceab5e34458824f36fac446fb665","_number":669096,"virtual_id_number":669096,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-07-04 12:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-07-04 11:42:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-07-04 11:42:13.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 11:42:13.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-07-04 12:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a25c791c9281a608db20b34ce8225ce55e22bb5","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 08:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ee9f49ffe124d98a46dc6d527bc5fa91fe3142e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/669096/1/check/openstack-tox-cover/f84a2d5/cover/ : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/669096/1/check/openstack-tox-lower-constraints/1bd8195/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/96/669096/1/check/openstack-tox-pep8/a427ff6/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/96/669096/1/check/openstack-tox-py27/f86dc56/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/96/669096/1/check/openstack-tox-py36/86e9212/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/96/669096/1/check/openstack-tox-py37/e722485/ : SUCCESS in 4m 14s\n- requirements-check http://logs.openstack.org/96/669096/1/check/requirements-check/1d711ad/ : SUCCESS in 2m 21s\n- build-openstack-releasenotes http://logs.openstack.org/96/669096/1/check/build-openstack-releasenotes/502b718/html/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/96/669096/1/check/openstack-tox-docs/eafe94e/html/ : SUCCESS in 2m 17s\n- openstackclient-check-plugins http://logs.openstack.org/96/669096/1/check/openstackclient-check-plugins/15fff0b/ : SUCCESS in 3m 50s (non-voting)\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/96/669096/1/check/muranoclient-functional-test-mysql-backend/a26c6d4/ : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6a49b5012accbfc935311900adde5e54f4c3aa84","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-07-04 11:42:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da1a9f97746a5cbb54c05c574d8ab82059bb75c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:42:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15ef48c09132ca9048205ebf39a17a00076e32fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:26:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5858c6ddd3792a0076394bdca2d3e9bfe166df13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:26:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/96/669096/1/gate/openstack-tox-lower-constraints/ea56a44/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/96/669096/1/gate/openstack-tox-pep8/ad7d28a/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/96/669096/1/gate/openstack-tox-py27/927d753/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/96/669096/1/gate/openstack-tox-py36/a9255ef/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/96/669096/1/gate/openstack-tox-py37/b83fdcc/ : SUCCESS in 3m 51s\n- requirements-check http://logs.openstack.org/96/669096/1/gate/requirements-check/5dd81a1/ : SUCCESS in 2m 15s\n- build-openstack-releasenotes http://logs.openstack.org/96/669096/1/gate/build-openstack-releasenotes/031b9e1/html/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/96/669096/1/gate/openstack-tox-docs/030135c/html/ : SUCCESS in 2m 37s\n- muranoclient-functional-test-mysql-backend http://logs.openstack.org/96/669096/1/gate/muranoclient-functional-test-mysql-backend/96c3d65/ : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3f4d15e14bc51af2cd299ade7f95231bc406f6c","revisions":{"b3f4d15e14bc51af2cd299ade7f95231bc406f6c":{"kind":"REWORK","_number":1,"created":"2019-07-04 08:21:13.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/96/669096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-muranoclient","ref":"refs/changes/96/669096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/96/669096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/96/669096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-muranoclient refs/changes/96/669096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-muranoclient refs/changes/96/669096/1"}}},"commit":{"parents":[{"commit":"e4db579c4be4a94c74a958857b33fa1d5d972f6c","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/e4db579c4be4a94c74a958857b33fa1d5d972f6c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:20:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:20:40.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: If2ca7919c0a4a488b2719e4063a2c700f23d3e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/b3f4d15e14bc51af2cd299ade7f95231bc406f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-muranoclient/commit/b3f4d15e14bc51af2cd299ade7f95231bc406f6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
