)]}'
{"id":"openstack%2Fpython-novaclient~1002874","triplet_id":"openstack%2Fpython-novaclient~master~Ide8cb232c5d50dc82a4e9709ed598a458e6264e5","project":"openstack/python-novaclient","branch":"master","attention_set":{"39263":{"account":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"last_update":"2026-08-31 18:33:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-31 18:33:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ide8cb232c5d50dc82a4e9709ed598a458e6264e5","subject":"Fixed max-servers handling in host-evacuate-live","status":"NEW","created":"2026-08-29 16:47:03.000000000","updated":"2026-09-05 09:14:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79b500884ca8ac7519148ebec4ddd198a566dcf9","_number":1002874,"virtual_id_number":1002874,"owner":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-29 18:27:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-08-29 16:55:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-29 16:55:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-08-29 18:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bda55abf248c38c59bbc5e783075eae1655f7039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"date":"2026-08-29 16:47:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31c8d35543226a02b409a2533a2b29c46c37f573","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 16:55:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dcc26456cfae579daac8d01cd884b70afd3dcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 18:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3742ff8db954aac82e625e1ccc49061\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d4687865b854e84a6c7b4a2c4e652f3 : SUCCESS in 1h 35m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7333b86b62ce428ea6a2b92e32f46cd4 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cdd6ea532341e59ed149b17883d699 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62f71d0a6f9246c0934a4c207090e75c : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97cf87941a5948b98b3bc80a640a3ba2 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/009fe982483f446d8a216bace6963c28 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5084963c64934b5dbac61f21dd89f7d2 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/450eef0ef6fa43dfb80191f67eac49ba : SUCCESS in 5m 14s\n- python-novaclient-functional https://zuul.opendev.org/t/openstack/build/97d8695741d44bfcac1fb1758cef97e9 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"44e3a885837a3ca131d8bbd278db9d6fbf62feb9","author":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"date":"2026-08-30 01:53:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bab2f56e48f79ed6887dd01863f0ec397a3f1e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 18:33:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79b500884ca8ac7519148ebec4ddd198a566dcf9","author":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"date":"2026-09-05 09:14:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff1bb274c6091fe57aec63bd1cc750146094b732","revisions":{"ff1bb274c6091fe57aec63bd1cc750146094b732":{"kind":"REWORK","_number":1,"created":"2026-08-29 16:47:03.000000000","uploader":{"_account_id":39263,"name":"Udayendu Kar","display_name":"Udayendu Kar","email":"udayendu.kar@gmail.com","username":"udayendu.kar"},"ref":"refs/changes/74/1002874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-novaclient","ref":"refs/changes/74/1002874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/74/1002874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/74/1002874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/74/1002874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-novaclient refs/changes/74/1002874/1"}}},"commit":{"parents":[{"commit":"3cdb7f59055f53ac6df58ce9b42481791a2fb741","subject":"bindep: Remove stale marker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/3cdb7f59055f53ac6df58ce9b42481791a2fb741"}]}],"author":{"name":"Udayendu Kar","email":"udayendu.kar@gmail.com","date":"2026-08-29 16:46:26.000000000","tz":330},"committer":{"name":"Udayendu Kar","email":"udayendu.kar@gmail.com","date":"2026-08-29 16:46:26.000000000","tz":330},"subject":"Fixed max-servers handling in host-evacuate-live","message":"Fixed max-servers handling in host-evacuate-live\n\nThe --max-servers option is intended to limit the number of\nsimultaneous live migrations during host evacuation. But previously,\nthe command stopped after submitting max-servers migrations.\n\nMaintain a concurrency window by waiting for active migrations\nto complete before submitting additional migrations, and continue\nuntil all servers on the source host have been evacuated.\n\nAdded unit tests covering invalid, single-migration, and\nmulti-migration concurrency limits.\n\nCloses-Bug: #1834482\n\nChange-Id: Ide8cb232c5d50dc82a4e9709ed598a458e6264e5\nSigned-off-by: Udayendu Kar \u003cudayendu.kar@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/ff1bb274c6091fe57aec63bd1cc750146094b732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/ff1bb274c6091fe57aec63bd1cc750146094b732"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
