)]}'
{"id":"openstack%2Fpython-novaclient~739940","triplet_id":"openstack%2Fpython-novaclient~master~I0e2c1895e50bc213f5346d8dc842c47519bacb99","project":"openstack/python-novaclient","branch":"master","hashtags":[],"change_id":"I0e2c1895e50bc213f5346d8dc842c47519bacb99","subject":"Add ensure-tox to novaclient-dsvm-functional","status":"ABANDONED","created":"2020-07-08 09:27:58.000000000","updated":"2020-07-09 17:01:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6c531c49537ce2d87cafa6b6a078463cad0b534","_number":739940,"virtual_id_number":739940,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-08 11:02:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-07-08 09:30:10.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-07-08 13:46:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 09:30:10.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-07-08 11:02:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 13:46:35.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"}],"messages":[{"id":"8d661fa84f05efbfc74ec2e0c688b6a285505fdc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-08 09:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2170afeadb255f1420db75fbcf9963ec38206d5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-08 09:30:10.000000000","message":"Patch Set 1:\n\nLet\u0027s see if https://review.opendev.org/#/c/725163/ is ready to go or not. If not then we can merge your fix to the legacy job.","accounts_in_message":[],"_revision_number":1},{"id":"f56704a918f4a765eec29a81c189389ed5efe86e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:02:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d62e3db3308d48f4a6b6de4774163841 : SUCCESS in 1h 23m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc345a942b444baa91933a7011c1d9f : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a3c8fbf110c4ca7bf7f82c18c988173 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69846e9edeaf4a12866d5faa27d2213d : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0b3382c73834e8eb0d72bdac701b635 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f19d3b297d304c968aa30e15ee9f7aed : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ac07ded2084bbcacb8c5f59508ef75 : SUCCESS in 9m 16s\n- novaclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1c8c658c62b4c0892a7813afa7c80f0 : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c6c531c49537ce2d87cafa6b6a078463cad0b534","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-09 17:01:57.000000000","message":"Abandoned\n\nFixed with https://review.opendev.org/#/c/725163/ instead.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f329f4fa780fdaed4256be8f10a61aa1951ad127","revisions":{"f329f4fa780fdaed4256be8f10a61aa1951ad127":{"kind":"REWORK","_number":1,"created":"2020-07-08 09:27:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/739940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-novaclient","ref":"refs/changes/40/739940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/40/739940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/40/739940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-novaclient refs/changes/40/739940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-novaclient refs/changes/40/739940/1"}}},"commit":{"parents":[{"commit":"600384fa3b046f3c01e7f375f384610df99fae90","subject":"Remove Babel requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/600384fa3b046f3c01e7f375f384610df99fae90"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-08 09:21:16.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-08 09:27:29.000000000","tz":120},"subject":"Add ensure-tox to novaclient-dsvm-functional","message":"Add ensure-tox to novaclient-dsvm-functional\n\nWith the new base image tox is not installed anymore by defualt, thus\nfunctional test job started to fail with tox command not found:\n\n[..]/post_test_hook.sh:main:L40: sudo -E -H -u zuul tox -e functional\nsudo: tox: command not found\n\nThis patch adds ensure-tox role to functional job\u0027s playbook.\n\nChange-Id: I0e2c1895e50bc213f5346d8dc842c47519bacb99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/f329f4fa780fdaed4256be8f10a61aa1951ad127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-novaclient/commit/f329f4fa780fdaed4256be8f10a61aa1951ad127"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
