)]}'
{"id":"openstack%2Fpython-observabilityclient~1003447","triplet_id":"openstack%2Fpython-observabilityclient~stable%2F2026.2~Ibe8a71e777389889becf78b7c4288fe12ae7b5d3","project":"openstack/python-observabilityclient","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"Ibe8a71e777389889becf78b7c4288fe12ae7b5d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-02 10:43:04.000000000","updated":"2026-09-02 15:26:49.000000000","submitted":"2026-09-02 15:25:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003447-create-2026.2","meta_rev_id":"f2062ff64c2b05aca2210331cee7cd37a7993626","_number":1003447,"virtual_id_number":1003447,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 15:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-02 15:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-02 15:01:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 10:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 15:01:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"14ca1c9a3afbba7296b5d6eca233c0d318cb1659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-02 10:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635729f00374ab12847b6b92caba783f34aa5ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 10:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a4a46e31004824b9164af35663aea5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ccc54a87ee8473ca6027a643f86f719 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4536bdc7ec9464eacdd56e8cabb1901 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68dc5a1c6e654d42a9bdd90b0067da29 : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/216dbf37bca844f78044d3c949f368e5 : SUCCESS in 8m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9d35b922aa7848c88df66add4ed5a9a9 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62208b488630418c9f992ddb36ef946e : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20efee52ed34e40867e56504e9e4799 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ba84f550b2ea0f1a32cd37394319f7c808ddef2a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 15:01:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"572570f65e68dad36fc28819558141e768ebf4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:01:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f2e93b0bc9ddd59f4e019c009527aedb5a9b640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:25:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2667f9465e040d0a8dbc0b8ee85f016\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a28fdca2fa64bbbbde3c35f1dc1999c : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a73f7b19d31d4cf291903fe0a44edd06 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37688b86502b44ed8dba54f0d8ca91fa : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/694acea54c2c4910b242c1d4f291eda6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416a9701195c4823a1a9b6382a55f624 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c81ae1e04007470b8a96b30318257bd2 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5e918acd635ecf70a4410622160849a9bd36206e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:25:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2062ff64c2b05aca2210331cee7cd37a7993626","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:26:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d5a15796394969bd90c9e50cad9957\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d692175dfd34cbbadf432c3520b8513 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2b33d95b205437cab3a77eb72dcc9f0 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f8add2ff606153debfb1d6e5b0b74a1188fe690","revisions":{"3f8add2ff606153debfb1d6e5b0b74a1188fe690":{"kind":"REWORK","_number":1,"created":"2026-09-02 10:43:04.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/47/1003447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-observabilityclient","ref":"refs/changes/47/1003447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/47/1003447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/47/1003447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/47/1003447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-observabilityclient refs/changes/47/1003447/1"}}},"commit":{"parents":[{"commit":"227f4a68374740e6cb621457620f22d980b589bd","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/227f4a68374740e6cb621457620f22d980b589bd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-02 10:43:02.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-02 10:43:02.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ibe8a71e777389889becf78b7c4288fe12ae7b5d3\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/3f8add2ff606153debfb1d6e5b0b74a1188fe690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/3f8add2ff606153debfb1d6e5b0b74a1188fe690"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
