)]}'
{"id":"openstack%2Fpython-observabilityclient~915395","triplet_id":"openstack%2Fpython-observabilityclient~master~Ie2db01d02483cf6767d5d54159ebf203e20dd2c5","project":"openstack/python-observabilityclient","branch":"master","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-04-10 11:52:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32240\u003e replied on the change","reason_account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-04-10 13:12:56.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-04-10 07:55:04.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-04-10 13:12:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2db01d02483cf6767d5d54159ebf203e20dd2c5","subject":"Sort column order in cli output","status":"MERGED","created":"2024-04-10 06:15:28.000000000","updated":"2024-04-10 13:12:56.000000000","submitted":"2024-04-10 13:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915395","meta_rev_id":"9ef4a034895506b34cb818ddc35a6e68f7ebda6b","_number":915395,"virtual_id_number":915395,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-10 13:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":2,"date":"2024-04-10 11:52:02.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":2,"date":"2024-04-10 07:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"all":[{"value":1,"date":"2024-04-10 11:52:02.000000000","_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 07:11:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 07:54:29.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-04-10 07:54:29.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-04-10 07:54:29.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"0e9ae3eb228f1afd01a1f6b730ae32ba82a9583c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-04-10 06:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e21c45b0ed0baa23c6cd6a914205d5d9b704c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 07:11:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fae9557b15c4087a7519f5d41b32128\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7fa17c9384047c1b619de046e8c00e5 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70afef4faac44689b1140d68d90c87a0 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11eb8da216544f94add7743969808aca : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f560f9e93a564b60a53dda7ddfc21d38 : SUCCESS in 6m 11s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/de1b8f75da7b4be3876c94ff28a1dfab : SUCCESS in 32m 33s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4d2d6c9e73b642609d3d03e26e34c686 : SUCCESS in 50m 51s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/46e3cd615eb74169a7147558706f0b33 : SUCCESS in 34m 36s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fb0beebf48a44010b0d1b7746a86d692 : SUCCESS in 37m 46s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fd85f66afc8473a99c3bd44383c2d94 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7ca5afde50bf9a50fa4d7f382cb90dadb5568c0e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-10 07:55:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1a59e4449fbd84c429021d76b58d5ea510cfe19","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-10 07:55:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b6d4e6a0f3d2e299f016ca22d91f35897293e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 09:07:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/762a57d3fb024906a2ddfb415f664458\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1a580bf29743e38d4a461f146ca489 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ea0446bb9d2492eb2157d89a7543bc9 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b75dd6fb6b674076b9efde180d6562d6 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/deeabb6e27334019a9055c0c2bd1dd5a : SUCCESS in 6m 02s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e7beee6dd5f744cba6c7ae22f950177b : SUCCESS in 59m 31s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f5ff040a69ab4de2a00c5b80c22b6766 : SUCCESS in 52m 48s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/013dfadde2f3405095b7f8503a2f00c0 : SUCCESS in 39m 27s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/ee00fcf727114583afb7a5af4305b92f : SUCCESS in 43m 38s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/d69804caa0af47f29de64e5abafb851c : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e41901b8dcf9fecf1a0adfba39993560e936e135","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2024-04-10 11:52:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a93173eaa721a7a65b4c22988d5581534d7482a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:52:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09082c770f020516beef1507c14a0f1a9ec8104a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 13:12:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c901bdf8b5c43aead88f93c9414bffe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ba8db5e8a04691ae1b109fa275d9c7 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5656b5db612446acad11c1131a165346 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d4439ebccbe460abe6053e6d274280c : SUCCESS in 3m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8e14ec7354dc41e194fa07f266182ce9 : SUCCESS in 52m 41s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ec6b12cbde1f4e509f906381dcacbf14 : SUCCESS in 52m 55s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6a9186f9422e49a0bf3bb55729602d8a : SUCCESS in 55m 49s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/28f51c71b6a8401a972ac427d974243a : SUCCESS in 1h 01m 45s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/4539fc2ddf39477588905d60dda8e1a6 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9ef4a034895506b34cb818ddc35a6e68f7ebda6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 13:12:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dcd00ad334b51a10a36df080cb2b1cf59d63ce3","revisions":{"3dcd00ad334b51a10a36df080cb2b1cf59d63ce3":{"kind":"REWORK","_number":1,"created":"2024-04-10 06:15:28.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/95/915395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-observabilityclient","ref":"refs/changes/95/915395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/95/915395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/95/915395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/95/915395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-observabilityclient refs/changes/95/915395/1"}}},"commit":{"parents":[{"commit":"ed042e2b16eda12805afff9c11d209eb59b7c1ae","subject":"Fix table formatter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/ed042e2b16eda12805afff9c11d209eb59b7c1ae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-04-10 06:13:13.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-04-10 06:13:13.000000000","tz":-240},"subject":"Sort column order in cli output","message":"Sort column order in cli output\n\nChange-Id: Ie2db01d02483cf6767d5d54159ebf203e20dd2c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/3dcd00ad334b51a10a36df080cb2b1cf59d63ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/3dcd00ad334b51a10a36df080cb2b1cf59d63ce3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
