)]}'
{"id":"openstack%2Fpython-observabilityclient~995110","triplet_id":"openstack%2Fpython-observabilityclient~master~I423f1f444c4d5ac4357db6e3c57c0e486c61c79d","project":"openstack/python-observabilityclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-08-10 15:07:12.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-08-10 13:48:13.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-07-28 17:05:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13177\u003e replied on the change","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-08-10 09:05:50.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"I423f1f444c4d5ac4357db6e3c57c0e486c61c79d","subject":"Keep session transport options","status":"MERGED","created":"2026-06-26 16:28:43.000000000","updated":"2026-08-10 15:08:09.000000000","submitted":"2026-08-10 15:07:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995110","meta_rev_id":"9e6374a48c9752917275fc45cc1e8832c259fa6a","_number":995110,"virtual_id_number":995110,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-10 15:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2026-08-10 13:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":1,"date":"2026-08-10 09:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-10 09:05:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2026-08-10 13:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 16:34:48.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2026-06-26 16:34:48.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2026-06-26 16:55:47.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2026-06-26 17:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-10 09:05:50.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"c1b173a6cbf48092edb5799714e62b11e6d3785a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-06-26 16:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"410265aa92d3f6af6dd091116f3744b434f2a3c8","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-06-26 16:55:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"08d6c786278a9a286d69684c5fd37de8243587f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 17:37:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba5a6d0ff5f46ef9c896287f96171e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a285641256e14cd4a32bccb18e411651 : SUCCESS in 2m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9e8c8648b844e91bc573a333ef174ca : FAILURE in 6m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13bd9748e78a4eddb844d4d92344ccb5 : FAILURE in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c0f8e2e9baf4bfab0e5a8ceb2b41b66 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae6c246ddbd44b59756bb6c14d186fe : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e0ed414fa4a4fac8b9f83209f71624f : SUCCESS in 3m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f163948aab874030b915851477fee290 : SUCCESS in 40m 26s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/62b195c5505a42029b037d38da15ebf1 : SUCCESS in 41m 50s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/329a651223d741e8b58c4a1bb12ca700 : SUCCESS in 48m 42s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/df983a310b7742828eb50399250e4e9e : SUCCESS in 58m 44s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/980d5b003a1a4f3ab89b95a44d5f6c7b : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"acd58b7d4b73c4a8176cef19cc1c9b7ed5e8a42a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-07-28 16:40:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"19bb6c3c67899aa4f0dd2efeb197287c29fc2ba1","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-07-28 16:41:59.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"48aa503f896f56b3b4a9da39f7bad88746e62419","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-28 17:05:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a15273c43b4647f826b588022f25514fbf60d6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 17:52:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35c75b3504a44d989e923bb77aa8cf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c45be9e8a04888a1ff6b173f581af7 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e66b1d8dc3f947729489f1a5dc138fb7 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f2e13b5fa4b4b8e953ce91625e1b875 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15ff434e50d04763a6c3bcf82a749b1c : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872abd13d7e84341ae172aa6de03193b : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5cbcb5f2e0d450dabc974c8ce3377c9 : SUCCESS in 3m 00s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9514f52b949544c683f3b97d1fb29e42 : SUCCESS in 42m 08s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee4a2f60da1481f86e0c170822e9f67 : SUCCESS in 51m 26s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a8a79d68cc0c4f19b263a52586addc9b : SUCCESS in 1h 00m 18s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/19457800cf9849a1a87cc46da152e759 : SUCCESS in 1h 05m 59s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/764ff47f57964138b492707bf53ccc27 : SUCCESS in 45m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"ec6b541b7ca21ebfd853e2882442871f443d4526","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-30 09:49:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ccd7fd8faa378026c666b0c3f00bf23671adfc9c","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-10 09:03:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"430f71ed63d7d8ecfbf058c9a5585f017612da74","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-10 09:03:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bd55c729fe0bf237636428e74eb2f4853c447ba","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-10 09:05:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4e98cc4584c2a740b6db988d5712678011e76641","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-08-10 13:48:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1f0916e69ad184628ff2f086e3cd8e91075e775e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 13:49:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1dca143921bc1a03b034bde9c5b12712d0b7b662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:07:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1faaa6ee20b42f19233789689825e80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3d29ab9cb614250abe7aaf416d09a85 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47ba67028d5149fd95840fd5f4f081d0 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7573d55f2209409194bc920a1cd019f1 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb7132f9c344f84814251727bc3dcd5 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6398c36557f94e0396597cd09bb45586 : SUCCESS in 3m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6ed753f09b2b49a5adca4835f7dbfcb4 : SUCCESS in 30m 27s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d79eaacb54fa40cdb1b190e16ef423a8 : SUCCESS in 58m 58s (non-voting)\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ba1fd8fb29a64a828f7bbbbc91114112 : SUCCESS in 1h 04m 05s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/d49038e8e23c4684874396501f48ee74 : SUCCESS in 1h 13m 19s (non-voting)\n- observabilityclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/0218bc3b06ed439babe15751e2e64cc6 : SUCCESS in 34m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"36e646f052e849a4ec1429ecac5c4d5540cab470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:07:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e6374a48c9752917275fc45cc1e8832c259fa6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:08:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8cc7336ed9244d1a034bd58b232754b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a4765424d8410d9762b50be1466aaa : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/970f6598c1ff4f84bd3d3eb0a7661270 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40977496240a3ce9ef20fef9d88161d7d470ce3b","revisions":{"84e1af7ae0ad6ba22e66288962dfa9b61c02e739":{"kind":"REWORK","_number":1,"created":"2026-06-26 16:28:43.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/10/995110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-observabilityclient","ref":"refs/changes/10/995110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/1"}}},"commit":{"parents":[{"commit":"d0614a7c36f5f290451c776dc4228b2318484e01","subject":"Merge \"tox: Use new constraints option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/d0614a7c36f5f290451c776dc4228b2318484e01"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-06-26 15:43:17.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-06-26 16:28:24.000000000","tz":180},"subject":"Keep session transport options","message":"Keep session transport options\n\ncurrently there\u0027s no way to use observabilityclient against Keystone\nendpoint (Aetos) that uses self-signed TLS cert,\nas verify/insecure option set by clouds.yaml/openstackclient\nis effectively overridden based solely on endpoint URL scheme.\n\nInstead, presume that all transport options are already properly set,\nand do not touch them at all when working with pre-created session.\n\nChange-Id: I423f1f444c4d5ac4357db6e3c57c0e486c61c79d\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/84e1af7ae0ad6ba22e66288962dfa9b61c02e739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/84e1af7ae0ad6ba22e66288962dfa9b61c02e739"}]},"branch":"refs/heads/master"},"40977496240a3ce9ef20fef9d88161d7d470ce3b":{"kind":"REWORK","_number":2,"created":"2026-07-28 16:40:33.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/10/995110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-observabilityclient","ref":"refs/changes/10/995110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-observabilityclient refs/changes/10/995110/2"}}},"commit":{"parents":[{"commit":"d0614a7c36f5f290451c776dc4228b2318484e01","subject":"Merge \"tox: Use new constraints option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/d0614a7c36f5f290451c776dc4228b2318484e01"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-06-26 15:43:17.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-07-28 16:39:57.000000000","tz":180},"subject":"Keep session transport options","message":"Keep session transport options\n\ncurrently there\u0027s no way to use observabilityclient against Keystone\nendpoint (Aetos) that uses self-signed TLS cert,\nas verify/insecure option set by clouds.yaml/openstackclient\nis effectively overridden based solely on endpoint URL scheme.\n\nInstead, presume that all transport options are already properly set,\nand do not touch them at all when working with pre-created session.\n\nAssisted-By: Cursor + Grok 4.5 High\nChange-Id: I423f1f444c4d5ac4357db6e3c57c0e486c61c79d\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/40977496240a3ce9ef20fef9d88161d7d470ce3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-observabilityclient/commit/40977496240a3ce9ef20fef9d88161d7d470ce3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
