)]}'
{"id":"openstack%2Fpython-openstackclient~1003062","triplet_id":"openstack%2Fpython-openstackclient~master~Idfadc02ca4ab0ad54245a485ff0dec92fb18ebf7","project":"openstack/python-openstackclient","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-09-01 12:38:20.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}},"hashtags":[],"change_id":"Idfadc02ca4ab0ad54245a485ff0dec92fb18ebf7","subject":"bump minimum version of openstacksdk","status":"MERGED","created":"2026-08-31 14:14:51.000000000","updated":"2026-09-01 14:44:37.000000000","submitted":"2026-09-01 14:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003062","meta_rev_id":"612cd534ae49501dc42e88209922721180acfbca","_number":1003062,"virtual_id_number":1003062,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 14:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-31 14:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-31 14:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 14:35:21.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-31 14:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8adb963e4215c86459331b07c8c3b733b877bb8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-31 14:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8a9a0bb6f07c99408a0aa754c0586e578b7b26d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-31 14:35:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15568693e152832b5e95fcdc61d21c224528bc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 14:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1354c7f72ac54bdf9ab8b32bee125e87\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d7e61cacf5dc4618a5565d7ba7876530 : FAILURE in 5m 15s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d015e23b175f4a6dafd3e35b48c17164 : FAILURE in 3m 16s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/2dc0b4b5726a41be83f6696bdf24cfa6 : FAILURE in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2648ae808a4e4f3f9fd7f4ba7b23df95 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c406f4487e64f978e793ae1d8548e32 : FAILURE in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39558e6077b34f8fa6521e226e3492fd : FAILURE in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd01411cdc7549faac14f6156b53faf6 : FAILURE in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcef1fe9f28f4115a218db81f305e84c : FAILURE in 6m 14s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ba0f7b71837240d195716963dac4724f : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7b81bdd7314d32b8d7495bcf990683 : FAILURE in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c14830762164d2485d20d02748a6199 : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8c91f9f40d34317aee6c339a7d6cb8e : FAILURE in 3m 54s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/e3f3b015446d4361b71e0657fba21094 : FAILURE in 4m 20s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/8e3fadbcaf374cb49e869790a11bfa92 : POST_FAILURE in 10m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2fdc361f57ea437c86154be668582cc9 : FAILURE in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"03a1cb8ce6b1416a117f3de674f03667f4156823","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-01 01:56:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"379a5b84911e1eacc128215869a880f0246d7bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:40:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1727bb2637544f3b9731bca6766fe47f\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f30e9defe2884d78b6b59dc09f2f2441 : FAILURE in 4m 40s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d6917d3e4aac45f39114100cdd8d03c8 : FAILURE in 3m 39s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/27ef786c0ac145b1beebc497d75891ac : FAILURE in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a80eca320007476ebb824a18e985ca84 : FAILURE in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44221966d42b40feadcaf808c52525da : FAILURE in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd87d42554d43b09b83742825feff66 : FAILURE in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b89dd8323c8140ae8135a2f710358418 : FAILURE in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edd34000d4604235b7c686f7c9b45a94 : FAILURE in 5m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b27dcc3fde2d49e89a326b4fb8da0f75 : FAILURE in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e4e709ff684906b599359b9fb5b320 : SUCCESS in 9m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9000d7cb816b4e43b531bbffa32ca5f4 : SUCCESS in 2m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bea4114af234fa28ca9590e76421327 : SUCCESS in 8m 40s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a27dc8ef4e4849259db7938079b6bbe9 : FAILURE in 4m 46s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/61c493ffae0f43f4bc38a4314208dfa7 : POST_FAILURE in 9m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d74d638ba40349b6a124cb092e9b0410 : FAILURE in 29m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cca8f6d5bf036616d48886583e74a5fc93e9280a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-01 08:54:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"393ab78dbc0a71d5a096cd77ed057631c278c5a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:09:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8354a2d2c7d4a8fb2da4bb0e9403151\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/25c09d47ccb546e9baef7812b10fe0c6 : FAILURE in 5m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6e1cd6bbd9d44e6490f7831dd02b5e38 : FAILURE in 2m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/c14629347b0a426481e7b1d71a7fbd0a : FAILURE in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ab6315688443bda967b3dd69f2806d : FAILURE in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4faf72ef77974720bc46527fe78f213c : FAILURE in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67db1a27c4284538930f66fe7e0b4ce9 : FAILURE in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/771c8e9dc9e5456a9d6a3d5579af959a : FAILURE in 2m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/418fd967f2f94d0994eb788fe6059cd3 : FAILURE in 4m 50s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d42dbccdda2f46858b3019497247404a : FAILURE in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8baab772d7004929a862d9d3a7a2cc20 : SUCCESS in 6m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0626987b455e4126bb053c74882c3d40 : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629c1c4a1245430689100c9fa6af52b7 : SUCCESS in 9m 11s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/fc05efe584334ac2b97b09f532d686f0 : FAILURE in 4m 03s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/a1b248d3aa4e4cdaa66608ce135ce0d5 : POST_FAILURE in 10m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5814c75b3aa14c3c834aae2fd5e01706 : FAILURE in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cbdb869a1b1167e4d450e8db269933c3e0f653ce","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-01 12:38:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebc0a5dcca1c7123884a7d30c41e64227afa523b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:59:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/103b0af32bec413e85eb7218d08fbb4d\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/85e2212de13447feb47a259676108a5f : SUCCESS in 3m 11s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e14e4f4481db4711aee965fb9df7389f : SUCCESS in 2m 34s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d39bd207ebdc49c4b5c78b3f08d2f8d2 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39864e20605648c996851a945555bde3 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192464e5c3924f57a8e7a883d884df95 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47cd0b6793b8417982735f282959e6ff : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebdd330bcaf442e2af210acf288b9895 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7119bb00de804ce9be9033989cd85f8f : SUCCESS in 7m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/51ee817efef74ff68fea4fe99df3f41f : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2196dd094c5c4837ab7017653933fc18 : SUCCESS in 5m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1d01a429b6045b381eafa0f0a4239b8 : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3fe993bb6564e7f85623afdf0f81569 : SUCCESS in 9m 16s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/986e552a416047a4964d72afb65cd0f3 : FAILURE in 4m 05s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/ba74c4621ea34c6ba2a9af2d7e214b39 : SUCCESS in 46m 12s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f909176ff3094f1a9d07a9b00960c486 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d10137aacebabf4aa2760be694bb1525868da713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:59:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82e45dc0da3e47847016c5d8475237ccb6395788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff39cf9aec742cdabe74d862f73a136\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cf3ab23fa1ce412183b1cf71da1bba48 : SUCCESS in 7m 33s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/79451b2e02a6401c94121d3c1294228f : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc52582698b4fde8e7f4cae10216310 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20d0f0bbf1704a2ab97f3b7bc7215989 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32c4efdaa44d4d3aa8315b14a9324e20 : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00baf40cd0a04964801382c654d8ef64 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c64f296fcd4fa48dd45f401862c776 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a54ee75df0d74496b512396a02670c58 : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9065b6416044a1eb55d02fdcad78b55 : SUCCESS in 10m 21s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/9c8fce203b84418bbf859fe60e511396 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d76ece31078ba11b8cfa7d1aca910695edd62111","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"612cd534ae49501dc42e88209922721180acfbca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e99ec0df65245fd8dbdc4c15a4a61f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d68415638e4c61a57a5e250076e47e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d46ca8d4801ebd1a74abb9151023aa4b9b3405b","revisions":{"6d46ca8d4801ebd1a74abb9151023aa4b9b3405b":{"kind":"REWORK","_number":1,"created":"2026-08-31 14:14:51.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/62/1003062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/62/1003062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/62/1003062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/62/1003062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/62/1003062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/62/1003062/1"}}},"commit":{"parents":[{"commit":"926da721a80fba06a3dd39be06f0505305d29d75","subject":"Merge \"Add a \"cloud cache delete\" command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/926da721a80fba06a3dd39be06f0505305d29d75"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-31 14:14:03.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-08-31 14:14:42.000000000","tz":-300},"subject":"bump minimum version of openstacksdk","message":"bump minimum version of openstacksdk\n\nThis version should have been bumped with Ie6bd39f1a67aba8d59be4cbabcd3d2f7dde168b9.\n\nChange-Id: Idfadc02ca4ab0ad54245a485ff0dec92fb18ebf7\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6d46ca8d4801ebd1a74abb9151023aa4b9b3405b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6d46ca8d4801ebd1a74abb9151023aa4b9b3405b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
