)]}'
{"id":"openstack%2Fpython-openstackclient~1004414","triplet_id":"openstack%2Fpython-openstackclient~master~I0cdbfe45dbb6aa24eb06f902b7b790d3846f36cf","project":"openstack/python-openstackclient","branch":"master","topic":"bp/clean-prune-cache-api","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-09-07 13:16:43.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0cdbfe45dbb6aa24eb06f902b7b790d3846f36cf","subject":"Add cached image clean and prune commands","status":"NEW","created":"2026-09-07 12:26:49.000000000","updated":"2026-09-08 06:08:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8621b5869555841bc935bc45b97a874ea8b2fe9","_number":1004414,"virtual_id_number":1004414,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-08 06:08:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 13:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c33b2e31dd4635e4c7f291c3f1d2317d60aec4f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-09-07 12:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1841a80b65861b1654e848d35e90d026c520709b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce4e07b4aabc439cbc5dc480a7cd2328\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ddb00fdd4b4a4dbd89a8836127664d8e : SUCCESS in 2m 30s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/97bd21dee90a4e9591f966c455409817 : SUCCESS in 3m 40s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8c1ffc154c404744b87c6611383c4c9f : SUCCESS in 2m 52s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/5dfc724e76d4448cbdbaa3e17630078f : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6828f819e194810853ed68e2d7a8173 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188502e6abd6401180d6b65891776bad : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12568c185f5b4fd8829f672bcf7acf1a : FAILURE in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9541bc2aa8244349a90efdedeec2952 : FAILURE in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a201e31b31be4f2faa2e3c2b661c86b3 : FAILURE in 4m 43s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/630dc0be8ab7468a999e64070ed3d3ff : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613bc77beb0b47d69986d885a478cd0e : SUCCESS in 3m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d6f66964544427eaf482f71fbf1229f : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e855c905c204710b5e7834175ad6a5d : SUCCESS in 5m 01s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/22e1dd36d1a44e919fc8b4d208335dd8 : FAILURE in 4m 04s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/d701226655fa49ca980e0189aa57b871 : FAILURE in 46m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/70dfab8527d34dfe862e997e6b06401a : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"603d496afd22363e81b76342f7664a208f5faf89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-09-08 05:25:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d8621b5869555841bc935bc45b97a874ea8b2fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 06:08:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c1dd79dd6748f1b516196b5afef4c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc30ebc0b69a4ec6b1946b9f054f48ea : SUCCESS in 2m 42s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6f2d1deb96d441e583d1f198aa8b68fb : SUCCESS in 5m 51s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8ebbf70fdb7641c0a8ed38b5b0131eaf : SUCCESS in 4m 55s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/546f7fd14d554b548f73041ec551fa28 : SUCCESS in 7m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a7c8907f6d477da0f517844b1f83af : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd0f30747b44f2dbac82639200cb10f : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0612bb45de9749bf939db7a32c19478e : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c881022d6904accac3cb8792fc58ada : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39aabf916ae54fa486d178ef1b95b802 : SUCCESS in 7m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3b74bbceeb9449faaac7c9b2833fe227 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b23415115b4495db8d14e2ecdb76a39 : SUCCESS in 6m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/478ec0039fda48c1aa90687b09541f9a : SUCCESS in 1m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56ed115a463431289fcda63b984e8c3 : SUCCESS in 7m 31s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/ca26b647efe04e489caedf45b0c100e9 : FAILURE in 2m 47s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/816843c77cc64532b5fa21fa403a1145 : SUCCESS in 39m 44s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5bf903bf7a0e4cc09a78c30b424b09b7 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb2d35c7c8d6ea1d2dd08cdae1753069bc95e967","revisions":{"617fc4c77d3782103b0b15a9e23eb2b4d67ed41c":{"kind":"REWORK","_number":1,"created":"2026-09-07 12:26:49.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/14/1004414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/14/1004414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/1"}}},"commit":{"parents":[{"commit":"6ef4d932afaff1b43fda42a54dd430f51ffc9bda","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6ef4d932afaff1b43fda42a54dd430f51ffc9bda"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-09-07 11:47:12.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-09-07 12:26:47.000000000","tz":0},"subject":"Add cached image clean and prune commands","message":"Add cached image clean and prune commands\n\nDepends-On: https://review.opendev.org/c/openstack/openstacksdk/+/1003403\nRelated-Blueprint: clean-prune-cache-api\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nChange-Id: I0cdbfe45dbb6aa24eb06f902b7b790d3846f36cf\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/617fc4c77d3782103b0b15a9e23eb2b4d67ed41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/617fc4c77d3782103b0b15a9e23eb2b4d67ed41c"}]},"branch":"refs/heads/master"},"eb2d35c7c8d6ea1d2dd08cdae1753069bc95e967":{"kind":"REWORK","_number":2,"created":"2026-09-08 05:25:40.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/14/1004414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/14/1004414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/14/1004414/2"}}},"commit":{"parents":[{"commit":"6ef4d932afaff1b43fda42a54dd430f51ffc9bda","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6ef4d932afaff1b43fda42a54dd430f51ffc9bda"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-09-07 11:47:12.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-09-08 05:24:52.000000000","tz":0},"subject":"Add cached image clean and prune commands","message":"Add cached image clean and prune commands\n\nDepends-On: https://review.opendev.org/c/openstack/openstacksdk/+/1003403\nRelated-Blueprint: clean-prune-cache-api\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nChange-Id: I0cdbfe45dbb6aa24eb06f902b7b790d3846f36cf\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/eb2d35c7c8d6ea1d2dd08cdae1753069bc95e967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/eb2d35c7c8d6ea1d2dd08cdae1753069bc95e967"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
