)]}'
{"id":"openstack%2Fpython-openstackclient~1004923","triplet_id":"openstack%2Fpython-openstackclient~master~Ibc36d629a750dc7faee6e1d6708364b60e2fcb6a","project":"openstack/python-openstackclient","branch":"master","topic":"identity/remove-keystoneclient","attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2026-09-09 21:37:26.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibc36d629a750dc7faee6e1d6708364b60e2fcb6a","subject":"Identity: Remove keystoneclient for v3","status":"NEW","created":"2026-09-09 20:17:52.000000000","updated":"2026-09-09 21:37:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"298c8500a54bf7a64bf72376ef88e289f0b2b6d4","_number":1004923,"virtual_id_number":1004923,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-09 21:37:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 21:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c403531fe2fe7c2bacb3d3946a01f5e3cdb8ff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-09-09 20:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"298c8500a54bf7a64bf72376ef88e289f0b2b6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:37:26.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2f67c67f5243538181dac97ffa04f9\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9775b674271246e58f4b56ee09a2c58d : SUCCESS in 6m 15s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e9b70183822d4074a863397a4e487ffa : SUCCESS in 6m 24s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/bdffab8eab9140de81a2bd7c7571608a : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27ed7556129d4640b3e4e1aa62ea3c60 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ac498fb21a478495ebff2c189ea60b : FAILURE in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b00b1326f4943da933eabf3b45de7d3 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b366adc5a25c43668015ec1e8226f6fe : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8266913a9dfe46f6ac8582ef35b6c638 : SUCCESS in 7m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3c0f5acfa943437196dceb0aee41fc00 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a32c9002a67403e9612c2157714c382 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ba7a5f26974843bce5e80ba9a290df : SUCCESS in 8m 51s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/1daab192c05e40c09b9d019fe3b0d483 : FAILURE in 3m 36s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/19e6465d6eb0432293f7c627991d2d45 : FAILURE in 1h 11m 52s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/07c8cdf39b0f4adfba901cad0f75280f : FAILURE in 1h 14m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75cb7756144c42cb9a7b381379548d3e508e9bd4","revisions":{"75cb7756144c42cb9a7b381379548d3e508e9bd4":{"kind":"REWORK","_number":1,"created":"2026-09-09 20:17:52.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/23/1004923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/23/1004923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/23/1004923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/23/1004923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/23/1004923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/23/1004923/1"}}},"commit":{"parents":[{"commit":"f5cf1fb300ecc9f4ff4b990945c018066be2fea7","subject":"Migrate \u0027endpoint group\u0027 commands to SDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f5cf1fb300ecc9f4ff4b990945c018066be2fea7"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-09-09 20:16:08.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-09-09 20:17:34.000000000","tz":-420},"subject":"Identity: Remove keystoneclient for v3","message":"Identity: Remove keystoneclient for v3\n\nChange-Id: Ibc36d629a750dc7faee6e1d6708364b60e2fcb6a\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/75cb7756144c42cb9a7b381379548d3e508e9bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/75cb7756144c42cb9a7b381379548d3e508e9bd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
