)]}'
{"id":"openstack%2Fpython-openstackclient~1005826","triplet_id":"openstack%2Fpython-openstackclient~master~I090ceb4fd98043dc1c21ef8e249f3f0209ec0779","project":"openstack/python-openstackclient","branch":"master","attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-16 15:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I090ceb4fd98043dc1c21ef8e249f3f0209ec0779","subject":"Add router option to floating IP create and set commands","status":"NEW","created":"2026-09-16 04:04:59.000000000","updated":"2026-09-16 15:12:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7deedb6716eb6855b185c6e0fa4971be72bf9ee","_number":1005826,"virtual_id_number":1005826,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 05:11:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-16 15:12:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 05:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 15:12:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"e2e8994f7ede0d8bdc8e6944448650eb63ab22ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-16 04:04:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0428d22edf2a972a21615dcfdb2c382c2825971d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-16 04:08:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c3d117672e5d007de7c5fff3c161134eec91c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 05:11:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e58986f41cb491eb7a2e558038b6e73\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b7f3088ac7c842cb85da85558f13ad66 : SUCCESS in 5m 11s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ff158f2e6f2a4dc99eae88d72f47d1a4 : SUCCESS in 5m 29s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/a69891afe87649b98856b711be594c51 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/422a1c3a63154bf3814e132f1db0d92d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c401992fecf4bd282e346ba533f865b : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/115ca145e6f04557b19b8fc06d75efd5 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ea92081bfc24c4bb082a758ea1f790a : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/218f05d43e334b3f96454436ddb4a145 : SUCCESS in 4m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3df8e74356104e54b8ade31400a4c914 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2906f51c4a73438b815a065b2a763c9d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95dad49ef0d24a45aa607ceb4cabe214 : SUCCESS in 8m 38s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/e4855256f5dd42b5968ce6026e971191 : FAILURE in 6m 14s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/b7083ee0d8914dd6bec07a7d1f052103 : SUCCESS in 1h 00m 47s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1cb8186242234be1bd5e6b35c850071e : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c7deedb6716eb6855b185c6e0fa4971be72bf9ee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 15:12:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81ae5397e27236450db9348b59f1588f904583e9","revisions":{"8c229ce0589564567987c7811a8ad95aa2987e05":{"kind":"REWORK","_number":1,"created":"2026-09-16 04:04:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/26/1005826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/26/1005826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/1"}}},"commit":{"parents":[{"commit":"6682aa1d21a72b642ef88418461a2a630b448ff1","subject":"Merge \"Migrate \u0027endpoint group\u0027 commands to SDK\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6682aa1d21a72b642ef88418461a2a630b448ff1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-16 03:55:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-16 04:04:49.000000000","tz":120},"subject":"Add router option to floating IP create and set commands","message":"Add router option to floating IP create and set commands\n\nNeutron supports selecting the router that hosts a floating IP when the\nfloating-ip-router-writable extension is enabled. Expose this capability\nthrough the OpenStackClient floating IP commands.\n\nAdd a --router option to `network floating ip create` and\n`network floating ip set`.\n\nRelated-Bug: #2072505\n\nChange-Id: I090ceb4fd98043dc1c21ef8e249f3f0209ec0779\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8c229ce0589564567987c7811a8ad95aa2987e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8c229ce0589564567987c7811a8ad95aa2987e05"}]},"branch":"refs/heads/master"},"81ae5397e27236450db9348b59f1588f904583e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-16 04:08:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/26/1005826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/26/1005826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/26/1005826/2"}}},"commit":{"parents":[{"commit":"6682aa1d21a72b642ef88418461a2a630b448ff1","subject":"Merge \"Migrate \u0027endpoint group\u0027 commands to SDK\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6682aa1d21a72b642ef88418461a2a630b448ff1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-16 03:55:33.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-16 04:08:05.000000000","tz":120},"subject":"Add router option to floating IP create and set commands","message":"Add router option to floating IP create and set commands\n\nNeutron supports selecting the router that hosts a floating IP when the\nfloating-ip-router-writable extension is enabled. Expose this capability\nthrough the OpenStackClient floating IP commands.\n\nAdd a --router option to `network floating ip create` and\n`network floating ip set`.\n\nRelated-Bug: #2072505\n\nAssisted-by: GPT-5.6 Luna\n\nChange-Id: I090ceb4fd98043dc1c21ef8e249f3f0209ec0779\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/81ae5397e27236450db9348b59f1588f904583e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/81ae5397e27236450db9348b59f1588f904583e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
