)]}'
{"id":"openstack%2Fpython-openstackclient~1007600","triplet_id":"openstack%2Fpython-openstackclient~master~I0229208246d6eee5b563bf53651f2cb24e6dfdaa","project":"openstack/python-openstackclient","branch":"master","attention_set":{},"removed_from_attention_set":{"39381":{"account":{"_account_id":39381,"name":"taha hosseini","display_name":"t.hosseini","email":"taaahaaa75@gmail.com","username":"t.hosseini"},"last_update":"2026-09-28 13:39:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0229208246d6eee5b563bf53651f2cb24e6dfdaa","subject":"Normalize additional server list columns","status":"MERGED","created":"2026-09-26 19:17:14.000000000","updated":"2026-09-28 13:41:19.000000000","submitted":"2026-09-28 13:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007600","meta_rev_id":"4894618ba487eafc98a47443cbf8fae19707de58","_number":1007600,"virtual_id_number":1007600,"owner":{"_account_id":39381,"name":"taha hosseini","display_name":"t.hosseini","email":"taaahaaa75@gmail.com","username":"t.hosseini"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 13:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-28 12:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-28 12:21:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-26 20:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-28 12:21:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ff241eefdecca05484e927545fd88a8120f3524d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39381,"name":"taha hosseini","display_name":"t.hosseini","email":"taaahaaa75@gmail.com","username":"t.hosseini"},"date":"2026-09-26 19:17:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f412787eb21c2515a812b726f861f2ad95832d4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 20:39:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e889657973486baa425d6cdb67f130\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c3d27b2f09184d5e957f6b6e07f3b94e : SUCCESS in 5m 49s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b07c5d76881c429f96af71cf6f77d99a : SUCCESS in 3m 54s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/334e60b045e5435ea553724fa6cb4ace : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8da1de4cc97400c873a0ec79160b0a5 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9cc4c506b443dbb3a3cc6a1c3849bb : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c768fccce6c74ac689ab7f6cb31ca5e8 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3dbbd19234714ccebfbfe4a53b59f0cf : SUCCESS in 3m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/103e850d841948fba2f159054537b567 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2096f4c7424db686287f6ce2a8d66c : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b3c1784b14a45acb83ecda7c437e1fb : SUCCESS in 9m 28s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/f1c378433af544eab2d8dc83e76a86d7 : FAILURE in 4m 33s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/522b76b535594ac9a2deec4f4ddde6c9 : SUCCESS in 1h 12m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6bc61848a25d4092ba48e73b072f9589 : SUCCESS in 40m 52s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/5ab57ac8bd8b4201952c643412f50287 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":1},{"id":"46b0fb7e25c53f43ec8e3ed5d5b30132022ae9a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-28 12:21:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72e2351cab689e402a1854e878add651893227dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:23:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a62577ffc8809fc9eae0ac302ef9a1f8bd9a3bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:39:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5d1f5d1dc54499bb5f25d9f25faba9\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/70fa95e8cc994fa48c8a0a6807d5cda2 : SUCCESS in 5m 20s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/5e0c5a20dda64eb4beb66539b5b93705 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca24adf17834e409be75444876e9a55 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7676ec5acc7a4a37ba50d9e3e4837292 : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f3a0f2d6eef4246bae163cde194eb28 : SUCCESS in 6m 10s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e1ea70c5a66945468734bd797e8c83ac : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58b442c49944458af8ec99f3f9a0346 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ea244f9491e4d039d233169342f1e14 : SUCCESS in 7m 06s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/5c204577df8a45db9e3d566cb3093f3e : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0751483e296b3ee1ce31970510e8992bae81996b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:39:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4894618ba487eafc98a47443cbf8fae19707de58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8204e626c3a0466ab4a84775a5caf3d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a90eeb03ad4f409f072f5ee4d80859 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","revisions":{"8481e0cfb42237fa4f46ca711c27b3ddc612346e":{"kind":"REWORK","_number":1,"created":"2026-09-26 19:17:14.000000000","uploader":{"_account_id":39381,"name":"taha hosseini","display_name":"t.hosseini","email":"taaahaaa75@gmail.com","username":"t.hosseini"},"ref":"refs/changes/00/1007600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/00/1007600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/00/1007600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/00/1007600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/00/1007600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/00/1007600/1"}}},"commit":{"parents":[{"commit":"37b3b0363d52c5f72d94b58c43c35181c839834c","subject":"Merge \"Add functional tests and Zuul job for accelerator commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/37b3b0363d52c5f72d94b58c43c35181c839834c"}]}],"author":{"name":"t.hosseini","email":"taaahaaa75@gmail.com","date":"2026-09-26 19:16:54.000000000","tz":210},"committer":{"name":"t.hosseini","email":"taaahaaa75@gmail.com","date":"2026-09-26 19:16:54.000000000","tz":210},"subject":"Normalize additional server list columns","message":"Normalize additional server list columns\n\nServer list adds some columns on demand before Cliff filters the output. Normalize those requested column names so this path accepts the same case and underscore variants as Cliff.\n\nChange-Id: I0229208246d6eee5b563bf53651f2cb24e6dfdaa\nSigned-off-by: t.hosseini \u003ctaaahaaa75@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
