)]}'
{"id":"openstack%2Fpython-openstackclient~1008088","triplet_id":"openstack%2Fpython-openstackclient~master~I1232a1bc691a2f2e3c14659b5854ee9c15458755","project":"openstack/python-openstackclient","branch":"master","attention_set":{},"removed_from_attention_set":{"7654":{"account":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"last_update":"2026-10-05 10:07:58.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-10-02 16:44:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I1232a1bc691a2f2e3c14659b5854ee9c15458755","subject":"volume: Fix volume attachments list query options handling","status":"MERGED","created":"2026-09-30 15:53:56.000000000","updated":"2026-10-05 10:10:15.000000000","submitted":"2026-10-05 10:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008088","meta_rev_id":"2b8acb525184ffd00fdd120352cf1d793ef70589","_number":1008088,"virtual_id_number":1008088,"owner":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-05 10:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-10-05 08:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-10-05 08:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-30 16:06:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-09-30 17:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-01 07:47:17.000000000","updated_by":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"df06ee3856f98f0025f4de8eae99b24d657629a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-09-30 15:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee6127c186065788cbd1bf9789129324a5a7b004","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-30 16:06:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5fc57ca3e3c955b95dfde2d0d620689e9b678ce5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-30 16:11:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"046b2377d624a6db5e867562b3b8513f5524547c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 17:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a79e3f504f142cfa28893861bdb94e9\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f27adc6194cc499fb047d784187f9d0a : SUCCESS in 5m 38s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/191ad73b96b34d6ca9339716400da86a : FAILURE in 4m 08s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/657bc2c6fc7a408682f283dce3ee99ff : FAILURE in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e9198cf3c4d4574bab974b22bcc625f : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6445bfefe82d49e6851dc8275a14cf85 : FAILURE in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8ad535e1a7e4b4ea0226fb30805f019 : FAILURE in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a807ab5c1ccd43649b19d634e8a6e0b3 : FAILURE in 6m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/17dab14b8a7e47be89412e36bb7137f3 : FAILURE in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72469c68aba74781b3a522a8def13d00 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e40d1cdc9154f65adc768c1715e51a8 : SUCCESS in 8m 57s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/7fb29d0ea22442f690d8e038fa4450ad : FAILURE in 6m 12s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c37bd9c2b45c4f73bb1239a2f11091ae : FAILURE in 1h 16m 56s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a83c60a018604302ac74d006bc8757fe : FAILURE in 44m 36s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/73bbdbd6a25d422da7d15988e6e99bd5 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2f39b2f3c8db1d1727436d28d74851f8c4a9b6fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 07:06:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"104b8a24e323fe5982e32f624645a938fd660726","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 07:17:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e11a7fb5344c2fc40522bb9451dfbbff3ab01aa4","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 08:03:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fd971bf63a827d1005777075c80caa774058cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 08:23:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa89fcab2e32452aad545104d40f01b6\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d80810793c154e738de566878f65f2e6 : SUCCESS in 4m 59s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1dacd04f3fe7469fac6b7253c059d22e : FAILURE in 4m 20s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6ee38432031245d29cd1eaf69109744e : FAILURE in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d586b6ae838e46f4be4ad8346b1ecc45 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6083c15e5d349ee9eb0c21b76e68a35 : FAILURE in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6029a0c0ff9841d18193f457273e5286 : FAILURE in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/180cfdd8d36247f3bbc3b7422f52bfbf : FAILURE in 6m 06s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/f846fdaf892746318fe63430ba3a4c19 : FAILURE in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2b8350d0dd4af9994902e1a9999018 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020362d9afdf41fd9e0281c392008832 : SUCCESS in 4m 47s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/1298ff9536de4056a2547a58719f5bf6 : FAILURE in 4m 05s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/8eaf422e518449909bd0f05235303306 : FAILURE in 47m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a12f7ef7e4f941d6ba6c80d14a97bfe8 : FAILURE in 1h 12m 34s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/3c9552f1445f4cd789eb70b6731ad58e : SUCCESS in 32m 16s","accounts_in_message":[],"_revision_number":2},{"id":"352f9cb59cd97252d87d7e635ea35ba0101e5dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 08:27:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5f166f65da290caf955113b9e0349257b0e52ae9","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 08:30:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a16e3547d45cc59d5f720f487ce4c36df6e87cbe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 09:14:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f8910dad22bbd442fe0c6e8ada2ae8c60b48e2d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 09:42:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e81cf4dd8afe361122210e78b49b568c1eef02f2","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 09:52:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"30d9031bf7b45f2af63a551d33e980277789918a","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 10:16:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"561be2613c5bd51ad79bf4b5cf12cd9ffef863f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 10:39:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"59ece75a523e3a50d27201988b64c73c66931a77","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 10:42:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"eaa7ec9594216fca386a73e7f36df705e68756f1","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 10:47:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f59dd871a8236172685bbf059a243f1e19aeb7f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 10:59:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7af0b3597630e1b29de8b1ca5a33b1595fef56cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:26:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92287fcd4d5042518316e68ad8371619\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/22103dc4a8284e85a77b59773cd6fccf : SUCCESS in 6m 49s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8f5bb6165325453db4a76fdfc7d41a15 : FAILURE in 5m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/af62aad712ef458e835b76f95811ff76 : FAILURE in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67888d15b30a4330907306ea45b66d95 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cd2fd5df3c48338ecf214f7b5ddddc : FAILURE in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c8df37963d442f79f10b7926d9105c2 : FAILURE in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1685ee2a95b43d9a60bb5eb4843094c : FAILURE in 8m 26s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4985b919fbce44348a5cfda54f496c1b : FAILURE in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d89a5393ff47bf83378f79c90a7b35 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e78fe87ffd8f447ca84ee87401eeac92 : SUCCESS in 10m 41s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/b463bbf0a1d24a35b9f01b9a6a8c4871 : FAILURE in 4m 55s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/2926fb9e54e9436f927a2909757041b9 : FAILURE in 40m 09s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/20abc9e7e9ee40d9b4db4b9b9f2cd84f : FAILURE in 45m 41s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0ca1be0b3418459780e87d7d5e40c339 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":5},{"id":"c833d69e18fdfab92f75a17efa757a5e60962902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 13:12:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"3ec131d185194039372db87a9cc08e11734030e3","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 13:14:19.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"09d5d4fbabebd885d8fce658cf2c87d36de32137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:24:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b87bc05990244f38c7137e6966e51c4\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5566cbab1c4840778c09f9396a15c53c : SUCCESS in 5m 10s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fc6eaa14678045ff81c7ff86c496ab49 : SUCCESS in 5m 00s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4b4e3c3f0dca4ed48bbf8d82a10aeba7 : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/342e6dc54be644b994d850cc50792df7 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc109cc193b348fe95a04a79adfa84c6 : FAILURE in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ad34bb1a7834ef099957fc240b4f9ae : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0256e259c625473e8476669e5d91fa12 : SUCCESS in 5m 58s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8462bbb641a84427b6c832c8610e7df9 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c9511c57ed4dcb8ce309eded8ac241 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38312e0a92b45c3b5c83f4897c61ec9 : SUCCESS in 4m 40s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/0bcb820416b04bc6a0934c68b1404ab5 : FAILURE in 4m 55s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/428367d53a144c21b6a25f64e5d1cc5b : FAILURE in 1h 05m 19s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a7f9c1ce4e484ca09117cd8d29d0bfcb : FAILURE in 1h 08m 42s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/352106f27e534c67b7839135141a5270 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":6},{"id":"3509d1ffa80b0853f51647815ff6a5ab9bec0571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 16:45:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"dd64fd02bf66ac9251f1f39e24a73de804771209","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 16:48:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4028476c6fc3b140d7dfd72ebec8faabb8129cd3","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 16:49:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"32ad2ba7fb92c6929730190c083c63e7da2fa0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 18:04:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc73afe1354494bbc0a337da32437ca\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/eb77d9a06e8345e687cfca7a55eb8aea : SUCCESS in 5m 59s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ab98a4a8dc614dc7a329c7c34fc75def : SUCCESS in 4m 27s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9f43dd3106df4746adc229ad607914d7 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08fc28c4856e4f25a6d5b5445ea7f6af : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff68f4c7eba24fffb7dcec71e5a53122 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0b05afe615b419f8c564b39ed7974b4 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/189eb089949a4a398a2c2fade0ea472f : SUCCESS in 4m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/881cf826f0d046e89fea18b1e2190d8a : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95bcaa18460142a09a25c362e943e127 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b255ac4ca547af9d73957bc22b53ce : SUCCESS in 3m 58s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/4a2cebf268cb4fe691e71ba40110d1e7 : FAILURE in 4m 22s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/d8aa22e479d94406a35791560db68946 : FAILURE in 41m 44s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f0e791ff2b57400dbb8869781a3069fd : FAILURE in 1h 12m 40s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/13298755b1b14f2d86256456e09ee230 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c75b33aab1f05c8dcb8e864330543e14520da493","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 18:38:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3975dabcf06f8bb835411721611718d2970420e0","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 19:53:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6f712d40cd456ea69e51fff2bc979182203d7116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 21:02:50.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d7b66e28a1e4089975bd4f80707683f\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/2fe514a9c61e42f68abb1d381babd801 : SUCCESS in 3m 20s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6ff6fe77d3e842968a2d5884d4a0ae9f : SUCCESS in 4m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b8b122ad885a41899be40823bf4fe2aa : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ee6641499a4a49aa770912fa40ade2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7db6b158751498081e5cc87d448961c : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e515bd3c02e43a5821913e488ae8837 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/37d62b139ba44514892d2780148ecf2f : SUCCESS in 5m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5a38221686374a58accf86d8b375621d : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d239480aa6c48e69e0c1fb3f3639e6b : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f24c904dc24d44f6b604b2bc960fb9d0 : SUCCESS in 10m 35s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/62d3d1ca95e64deeb607cc2ad0504a70 : FAILURE in 4m 37s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/819586f7238142f78f7dbe41bf28f8f2 : FAILURE in 1h 06m 12s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2cd86ac5715f4b0c9545f9fa7b6997c8 : FAILURE in 1h 00m 17s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0889496ed3434330b5552c0d53145e64 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":7},{"id":"bdf3233f4b6619da60f9239293cad8afdb7552c4","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-01 23:48:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"337eb4c54fd9b2fabc1db1e0c51b5273592635ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 01:08:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dea084c3bea401fb08b8a6da0e86123\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b3eb13e5220244acb79c549a3668f32a : SUCCESS in 5m 41s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/12af9eab0559463abcbd2f5e2bb3f3bf : SUCCESS in 7m 19s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9f85225943ce417f9a669d3710f537fd : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/828912c356784e848ae5b7b89e29887d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4011498ccbc4dd293a4de7d42803342 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81b46c6f9b884babb33c3a2db26846af : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c33aa0ee2bc04c5ebc9ac1c08d6c8a1c : SUCCESS in 5m 19s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/66c030fb80cd4b4f9dd2188be610c814 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd59e63e7f245d59276cc38c8d09c95 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7e03280a0343fbbb392bda5c4dd833 : SUCCESS in 8m 30s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a40816438c4b4875af4f85717e9f3c5f : FAILURE in 2m 49s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/e2afcba7631d45628ba3fca537782212 : FAILURE in 1h 12m 45s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/42f2ac8270e9438cb723f51f6e264652 : FAILURE in 1h 14m 02s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/eab92e24d1b94836950dfce716cb12e9 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":7},{"id":"6e2dc79b7360ef895f3ed6f2ea344ef9e1308dc3","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-02 06:40:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"46161ebb693a097845ace4c4ea11bb820f6548cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 16:44:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8e7aeb47e5609e27d9c3059996c712e369150df6","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-02 23:50:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ca144ebb8002db12fd486b63566470d927424459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 01:09:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed002358641452d935aaef30d323ed4\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c9ee848abcc54adeab810444c83ab46c : SUCCESS in 4m 57s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3e55c0274f634e5a96ccc13662e31f69 : SUCCESS in 4m 37s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d3dde901f5524bd29d7c422a173dabd3 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d832c27cb61140c6b1219e47729fc80d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f854f9af468f4f81a608ffa86aa45533 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93daad34305b427b8e6f1db7276cbd8a : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e6d52167a7c40879db7613ecde88b23 : SUCCESS in 6m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/afd287ad6f6f42358b2296c7f38751ea : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11c676877964df88a80594615660cc7 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc64d3e9fe804c9289c8ff27fe1bd47a : SUCCESS in 8m 10s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/ff4feca2fa87449eac0798b0a953f667 : FAILURE in 3m 19s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/213abe2b73d6403fbea12ed77733d13c : SUCCESS in 1h 13m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e30d235aa898447fa00ecea73bf4813b : SUCCESS in 1h 13m 00s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/a4d0577e83d84380a499faa544e9f4b6 : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":7},{"id":"6688946fecc8dfffc646c9aef4f1b01ed4486a46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:06:41.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"18a73f70b40f1498824b8d8691864532f3ce7c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:07:55.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"7bda24f3e36d57bcb1cc023d747672a336060ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:08:29.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"314bd84ba7fc2efa6abbda01b29aeecef6c8ded7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:09:12.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"dec4dfc333a6afb57fa471e14ebe16e51941087a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:11:28.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"f738ddafe1d0b6c7fc47ab3a037da49c6fde273e","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-03 09:11:47.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9da928dd31db560ba9e564447a12f8dc05ac91cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 10:32:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a768c05c865344f78ce0969350846f6f\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/187a0ca99e7f47a5858ff3b3ff1edc8f : SUCCESS in 3m 20s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/735aa9d4f86d472e9cc986a3ded5ab59 : SUCCESS in 6m 21s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9e7d4918ec514be882da78af93169feb : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52ded7910224e029433717b20563cf1 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29cef07ed7644cf29f679946d83f7bac : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c6a98fbf88d4c8a9a7ecd64502cf6ac : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a032cc1c97fc471a922affaeab62fff8 : SUCCESS in 7m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d5005692ed6a412daeddeede91250f95 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4649ca4d4754e79943aa89889a2c99d : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ff67ee22ea74c5da0d3a54e449c75b8 : SUCCESS in 4m 12s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/ed9c2325538e4f419d07f2a101674521 : FAILURE in 3m 37s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/bd49c753028a4d50bed0a2ebe6cfd21a : SUCCESS in 1h 20m 28s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d64ecf3b79ee456d84fd5a3134a4a065 : SUCCESS in 56m 48s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/46c09c007c454bd28f4c55af46af8937 : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":12},{"id":"2935641607a27ec2084534a74d14a1c40102d3cb","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-10-05 07:09:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c4b59cd2421799c4fdcf63f296da8dfb58b6bfc3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-05 08:24:39.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"f039b787708de5f3d1792f51b86052ad0f907158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 08:25:09.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"cf866c1902b186d842136c57e77518fff32bf26c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 10:07:58.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7de70cda33294ac4b8f3b401ae62b63a\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/20020d0143fb43908646bbf4b214d47e : SUCCESS in 5m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d573967a1dbc4c7883ee5173108c52d2 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a104562e8f2c49a99a6885414b797696 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f91f45a6a8c649c29877c65846462175 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70c7729aebd241ec91bb884f77b277e6 : SUCCESS in 6m 48s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/dac70169d9ee4a3e9ef453a79705f277 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172b1a4bd26241598b1248526bf48fe6 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81b014a0d1b0434abd4375f645c5a016 : SUCCESS in 10m 31s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/0ec412c35aa24d5998863e520c47b5b2 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":12},{"id":"11794e0b9a6e820e025204f5b9197ebf4868e561","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 10:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"2b8acb525184ffd00fdd120352cf1d793ef70589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 10:10:15.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0f4fad979c4900943b7a42f1edeffd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e5a78280f240f9adad8400114c3c16 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e0425a5a57d674a852f0c36b4b705af02d55b61f","revisions":{"5005b8f4fd81af4b2c1b22ea0902004c032e219c":{"kind":"REWORK","_number":1,"created":"2026-09-30 15:53:56.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/1"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:52:39.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5005b8f4fd81af4b2c1b22ea0902004c032e219c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5005b8f4fd81af4b2c1b22ea0902004c032e219c"}]},"branch":"refs/heads/master"},"82ddf9924b924094cdc2a8c77180716d51a4a0bb":{"kind":"REWORK","_number":2,"created":"2026-10-01 07:06:56.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/2"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 07:05:22.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/82ddf9924b924094cdc2a8c77180716d51a4a0bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/82ddf9924b924094cdc2a8c77180716d51a4a0bb"}]},"branch":"refs/heads/master"},"5cc3cd4a617567e6c21e00475d2023e251101ab1":{"kind":"REWORK","_number":3,"created":"2026-10-01 08:27:31.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/3"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 08:27:17.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5cc3cd4a617567e6c21e00475d2023e251101ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5cc3cd4a617567e6c21e00475d2023e251101ab1"}]},"branch":"refs/heads/master"},"b19b115dfb333038cc15b0e225230bf6ded02861":{"kind":"REWORK","_number":4,"created":"2026-10-01 09:42:06.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/4"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 09:41:52.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b19b115dfb333038cc15b0e225230bf6ded02861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b19b115dfb333038cc15b0e225230bf6ded02861"}]},"branch":"refs/heads/master"},"a4cad3d8595fb8f396f4943ba0607f8a280b9ffc":{"kind":"REWORK","_number":5,"created":"2026-10-01 10:39:08.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/5"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 10:38:53.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a4cad3d8595fb8f396f4943ba0607f8a280b9ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a4cad3d8595fb8f396f4943ba0607f8a280b9ffc"}]},"branch":"refs/heads/master"},"1d344c4558786172d7365eb73a6684b4d9bc6e4e":{"kind":"REWORK","_number":6,"created":"2026-10-01 13:12:51.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/6"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 13:12:38.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1d344c4558786172d7365eb73a6684b4d9bc6e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1d344c4558786172d7365eb73a6684b4d9bc6e4e"}]},"branch":"refs/heads/master"},"e856924bc98d47f32d35149c70a6ae230540b894":{"kind":"REWORK","_number":7,"created":"2026-10-01 16:45:52.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/7"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-01 16:45:13.000000000","tz":120},"subject":"Implements volume attachments list query options.","message":"Implements volume attachments list query options.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e856924bc98d47f32d35149c70a6ae230540b894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e856924bc98d47f32d35149c70a6ae230540b894"}]},"branch":"refs/heads/master"},"706fe2a717a01730f771144780c8871b02b2b452":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-10-03 09:06:41.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/8"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-03 09:06:41.000000000","tz":0},"subject":"Corrects volume attachments list query options handling.","message":"Corrects volume attachments list query options handling.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nAdds the missing `--server-id` option.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/706fe2a717a01730f771144780c8871b02b2b452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/706fe2a717a01730f771144780c8871b02b2b452"}]},"branch":"refs/heads/master","description":"Edit commit message"},"743438a205d34ea95b872798a7a3a07ea9ce8886":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-10-03 09:07:55.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/9"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-03 09:07:55.000000000","tz":0},"subject":"volume: fix volume attachments list query options handling.","message":"volume: fix volume attachments list query options handling.\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nAdds the missing `--server-id` option.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/743438a205d34ea95b872798a7a3a07ea9ce8886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/743438a205d34ea95b872798a7a3a07ea9ce8886"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f868c42db3504f5874f2d839d2ae70d3e9079ae7":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-10-03 09:08:29.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/10"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-03 09:08:29.000000000","tz":0},"subject":"volume: Fix volume attachments list query options handling","message":"volume: Fix volume attachments list query options handling\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nAdds the missing `--server-id` option.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f868c42db3504f5874f2d839d2ae70d3e9079ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f868c42db3504f5874f2d839d2ae70d3e9079ae7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fdc1a1f148ecb2219e754311e63c3086a8d1c9ff":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-10-03 09:09:12.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/11"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-03 09:09:12.000000000","tz":0},"subject":"volume: Fix volume attachments list query options handling","message":"volume: Fix volume attachments list query options handling\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nAdd the missing `--server-id` option.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/fdc1a1f148ecb2219e754311e63c3086a8d1c9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/fdc1a1f148ecb2219e754311e63c3086a8d1c9ff"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e0425a5a57d674a852f0c36b4b705af02d55b61f":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-10-03 09:11:28.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/88/1008088/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/88/1008088/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/88/1008088/12"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-09-30 15:46:58.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-10-03 09:11:28.000000000","tz":0},"subject":"volume: Fix volume attachments list query options handling","message":"volume: Fix volume attachments list query options handling\n\nIt was previously passing all search options with\n\u0027search_options\u0027, but that doesn\u0027t work with the\nSKD.\n\nAdd the missing \u0027--server-id\u0027 option.\n\nTask: 55710\nStory: 2011888\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nChange-Id: I1232a1bc691a2f2e3c14659b5854ee9c15458755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e0425a5a57d674a852f0c36b4b705af02d55b61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e0425a5a57d674a852f0c36b4b705af02d55b61f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
