)]}'
{"id":"openstack%2Fpython-openstackclient~1008456","triplet_id":"openstack%2Fpython-openstackclient~master~I94d718d76fdf0bcd322ed007d1831ca9d279c249","project":"openstack/python-openstackclient","branch":"master","topic":"bug/2026547","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-10-03 11:38:22.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-10-03 11:38:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I94d718d76fdf0bcd322ed007d1831ca9d279c249","subject":"network: Drop address scope ``shared`` CLI options","status":"ABANDONED","created":"2026-10-02 13:05:23.000000000","updated":"2026-10-03 11:38:22.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"91c9e965ad4db243ce83e3a076b5499115ef48aa","_number":1008456,"virtual_id_number":1008456,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-02 16:58:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2026-10-02 16:04:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 13:06:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-10-02 13:11:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-10-02 14:44:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-10-02 15:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5453c881eb8c7bbd7d1a42174c3f802194d6c86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-10-02 13:05:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a880147848e9921d50fc86d6f3b689c7126bd91","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-02 13:11:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e930cf0ca7b96ecbecfd2ea84fa8a445a6cf159","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-10-02 15:18:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"124d44ec665fba341f63c5ee14e1fe7e8ed85964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:59:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b701f2d33ae4451be390459426a1651\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d78a1b184b8b4987b2b480eb15b01862 : SUCCESS in 6m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7bd63686bd7d4733935422fba9894cbd : FAILURE in 5m 26s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ee94dd1e466a4cb39aedd29e4caea803 : FAILURE in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc749615ad8a4d3c905cdaa8665d0b6a : FAILURE in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8b5e3a09cb4e9c98ddb56dc8c50b14 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43b950181a4a4733a221ed33f40c4a43 : FAILURE in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d6745f09b864faa81abfc36fb78ab08 : FAILURE in 7m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/638ca19097844049a40638898eb1ddb8 : FAILURE in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9e02c7dffb463f927e9d0644135c07 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0a215c8bbae4bde9fb185fc9042192e : SUCCESS in 5m 17s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/0fa930af607c4946b0d5da3a0f766920 : FAILURE in 4m 40s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/9e11c7405b61452c9491cf14b639b7b7 : SUCCESS in 1h 14m 45s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/73cf3f1890af4265a8c5d737fb61b504 : SUCCESS in 1h 08m 42s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/add0f6b9134c464780ae534c8ce8e8fc : SUCCESS in 37m 17s","accounts_in_message":[],"_revision_number":1},{"id":"31cfb0571661f7cbf36c924486f87da2d1f89390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 16:04:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a728876a7e0d82113cc74f7b6dab61644cdef666","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 16:05:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b77ec0a256255d15fef56b89449088a07df19ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:58:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aa3adafcfac40b2a7b59e9e41bcedb7\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7cfe7f4476a041ef9ed7764bc49eef23 : SUCCESS in 7m 53s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5303a22712db42c5af2bfd5d8cb2419b : FAILURE in 7m 20s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/83af91692129413e873485f9d09c2eb4 : FAILURE in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c47fef86371b4789930d444ad9ffa14a : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107b7d0821f848798435440f83c66a7b : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b8399f142e84ab29d2ba74c2a9a02ca : FAILURE in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e9177e010754366b83b8fd99d652701 : FAILURE in 4m 01s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8d1300a49ee54960a15b8f0f38685824 : FAILURE in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef9c245325f4ba2a61cfcfd40e44e20 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/461a5bdb03f540a9bb9db456c7583ea1 : SUCCESS in 10m 02s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a8bc1eb72b6a4183b5f0f7650b2c09f3 : FAILURE in 3m 08s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/e4407d7ef83045dbbecf7ccdc18b52d5 : FAILURE in 41m 42s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/51c49674957943d5bd8a4c1a691fa756 : SUCCESS in 52m 20s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0b474368515e4be589227cb273f1e44d : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":2},{"id":"91c9e965ad4db243ce83e3a076b5499115ef48aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-10-03 11:38:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"913cf4c166a07e87f6073f03606ee6405f6dd7d1","revisions":{"a18403f15eb67d6e2f3c2ee84c8da07507dafe93":{"kind":"REWORK","_number":1,"created":"2026-10-02 13:05:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/1008456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/56/1008456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/1"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-10-02 13:04:52.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-10-02 13:04:52.000000000","tz":120},"subject":"network: Drop address scope ``shared`` CLI options","message":"network: Drop address scope ``shared`` CLI options\n\nNeutron removed the ``shared`` attribute from address scopes [1].\nRemove ``--share`` / ``--no-share``, list column output, and\ndocumentation that referred to sharing via that field.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/887965\n\nRelated-Bug: #2026547\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I94d718d76fdf0bcd322ed007d1831ca9d279c249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a18403f15eb67d6e2f3c2ee84c8da07507dafe93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a18403f15eb67d6e2f3c2ee84c8da07507dafe93"}]},"branch":"refs/heads/master"},"913cf4c166a07e87f6073f03606ee6405f6dd7d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-10-02 16:04:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/1008456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/56/1008456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/56/1008456/2"}}},"commit":{"parents":[{"commit":"389591bc8daed6fd9e32e7d3325920dddcaf04a8","subject":"tests: Drop testing for address scope shared attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/389591bc8daed6fd9e32e7d3325920dddcaf04a8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-10-02 13:04:52.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 16:03:59.000000000","tz":60},"subject":"network: Drop address scope ``shared`` CLI options","message":"network: Drop address scope ``shared`` CLI options\n\nNeutron removed the ``shared`` attribute from address scopes [1].\nRemove ``--share`` / ``--no-share``, list column output, and\ndocumentation that referred to sharing via that field.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/887965\n\nRelated-Bug: #2026547\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I94d718d76fdf0bcd322ed007d1831ca9d279c249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/913cf4c166a07e87f6073f03606ee6405f6dd7d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/913cf4c166a07e87f6073f03606ee6405f6dd7d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
