)]}'
{"id":"openstack%2Fpython-openstackclient~1008496","triplet_id":"openstack%2Fpython-openstackclient~master~I58b74e15791e1079a84696f23fdf66e324fbdb08","project":"openstack/python-openstackclient","branch":"master","hashtags":[],"change_id":"I58b74e15791e1079a84696f23fdf66e324fbdb08","subject":"tests: Drop testing for address scope shared attribute","status":"MERGED","created":"2026-10-02 16:04:05.000000000","updated":"2026-10-02 19:03:25.000000000","submitted":"2026-10-02 19:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008496","meta_rev_id":"ed6354e7c9a1f84123410cf9db9db951f51822ec","_number":1008496,"virtual_id_number":1008496,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-02 19:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-10-02 16:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-10-02 16:04:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 16:04:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-10-02 17:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e73070587b8982ec51c133763d85d1fce1a2a589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 16:04:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cddf061191fdef16ac4ffdf6515242283e4baa2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 16:04:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d1ad6cf197f041b5a6ef5e92fec3f1e28df262a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 17:31:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb7753b1c95b4fa7be4fd2da412e8714\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1f795aa40f7d4e84811a37b1c1200449 : SUCCESS in 3m 17s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/8c19fb59111d49fea739ad913a93295c : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0fd19f14f14985892ec6a5ecfc7353 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc7dd8e5c114c96a68aa0ba76d05f12 : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/073c9a5f935a469182e2440f6a7e2327 : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c9e6761225b4cec822d0097a7519d9b : SUCCESS in 4m 15s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c61fe3700f014f4891c59f71e3dd8517 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0a6184ac6e4446a683d496c254934c : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbee7784d9b64d39b09e842dc80c69f2 : SUCCESS in 4m 07s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/fe4d36c6984c4bfa84de6a60dc2e582c : FAILURE in 6m 32s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/f39f48558e464ffe98c44007d0b3237c : SUCCESS in 43m 14s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/85ca664531a141c78e9dc0c66c566880 : SUCCESS in 1h 23m 30s\n- osc-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0ac008757eff4d6286b2108a67bb52f8 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fdc2802a4a488b623b6bc02f2d6044b6a573d729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 17:32:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8aff4bc8757648b21baab447668dceb063aa80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 19:01:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e6f88dbb334fd5a1213d7cfea62259\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/348dc43101614f1a8a2a6087a623bf89 : SUCCESS in 2m 33s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0ce3cab4b0ae4cf691439dd560accd4b : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288d74377a1e4083b708cf835eec57a9 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ee15e32bae0437eae94c6f0ebf66287 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43bfb380a1984176889f9e9b84784f06 : SUCCESS in 8m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/f7c88151e1be4f32aaecfe19d0105b94 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128ea644c62345d693b486cb715c03c8 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e600e10f604e13a7a2c20e42f4ac11 : SUCCESS in 9m 21s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c61a059245b34b95a08424c9ede3cd7c : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e81f8f0ae251c3c41fc232bc3d0954c52259ff3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 19:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ed6354e7c9a1f84123410cf9db9db951f51822ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 19:03:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca9a013197b64ccb85824e2252c4e89a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ff6a7cbe0648278ca03610d335d177 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"389591bc8daed6fd9e32e7d3325920dddcaf04a8","revisions":{"389591bc8daed6fd9e32e7d3325920dddcaf04a8":{"kind":"REWORK","_number":1,"created":"2026-10-02 16:04:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/1008496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/96/1008496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/96/1008496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/96/1008496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/96/1008496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/96/1008496/1"}}},"commit":{"parents":[{"commit":"8481e0cfb42237fa4f46ca711c27b3ddc612346e","subject":"Normalize additional server list columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8481e0cfb42237fa4f46ca711c27b3ddc612346e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-10-02 16:02:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 16:03:57.000000000","tz":60},"subject":"tests: Drop testing for address scope shared attribute","message":"tests: Drop testing for address scope shared attribute\n\nNeutron has removed the \u0027shared\u0027 attribute from address scopes [1]. We\nmay wish to drop this feature entirely, but for now drop the test that\nno longer functions.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/887965\n\nChange-Id: I58b74e15791e1079a84696f23fdf66e324fbdb08\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nAssisted-By: Composer 2.5\nRelated-Bug: #2026547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/389591bc8daed6fd9e32e7d3325920dddcaf04a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/389591bc8daed6fd9e32e7d3325920dddcaf04a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
