)]}'
{"id":"openstack%2Fpython-openstackclient~882548","triplet_id":"openstack%2Fpython-openstackclient~master~I2eeade1ce6653be8e9179ecc40105182c5ff5f16","project":"openstack/python-openstackclient","branch":"master","topic":"black","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-10 14:47:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2eeade1ce6653be8e9179ecc40105182c5ff5f16","subject":"Blacken openstackclient.identity","status":"MERGED","created":"2023-05-08 10:52:00.000000000","updated":"2023-05-10 14:47:24.000000000","submitted":"2023-05-10 14:47:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882548-black","meta_rev_id":"ca617fd9d811feaf35f68808ae88f2ab7c29eebf","_number":882548,"virtual_id_number":882548,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 14:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2023-05-10 13:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2023-05-10 13:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 12:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 13:35:20.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"707718a7383c6627e21a22f50a395536b5480284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-08 10:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40cd6d3ba942359b6c8c44138b3069c29f3870b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 12:51:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/206d13b509b543658bc68a12f34ab074\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1061056bdb2f4a9b83a28f6ba2659436 : SUCCESS in 4m 08s (non-voting)\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0e149c817081428da57dbcdae271bf7c : SUCCESS in 4m 41s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/c4e6001af2634387900e8f1a87de70c6 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d5be0e80b3b43049a3a6d9e9340f215 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c729bd7cd5fd40b987e1d003ec182f23 : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d6c634feb64b8098912d10d8cb8c8d : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae1963413574d52b54df281bf525505 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f42d1918f12476196ef26ace7dc9a74 : SUCCESS in 6m 37s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a97b331732444be3939936c6bdd47ecc : SUCCESS in 7m 04s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/14e92fd60f6c4c82bbf997ee81adacf2 : SUCCESS in 1h 00m 37s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f49b3300a4f042f3a607d22411584927 : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":1},{"id":"471aede69d23bb116236ded84059fee2453860c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-10 09:51:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e589c7b7190750f73a75cc3ececf0375b959c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:46:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dce86518cc545838f706b2ffc3da973\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e202057d4e3044e3b52cdf26597ffe45 : SUCCESS in 4m 28s (non-voting)\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9e81b59e46ae4eb690765808d0a082a3 : SUCCESS in 4m 59s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1561b6eee7ab4ba398a8439f538002f8 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19cc34d09ff44c86a081621ec16d5486 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092826a041dc452c92aadcdbc3f9a878 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21913c59d0d84889917b4b99a2a0dd99 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72068080e98b4d988631229346511429 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/500861d15d11456abffc54fd9640e980 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecee7a206448405095192a6fa2779595 : SUCCESS in 5m 46s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/4f8c3c48821c464b99883b5f495373a8 : SUCCESS in 5m 55s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/bd03a78140ee4c15b0b8bcd0efa23903 : SUCCESS in 44m 16s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e5647c2b980b43ae99ac529bca1048a9 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":2},{"id":"33f6c0b3964749d559cdda76106f9147be916d7f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-05-10 13:35:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b0a6896f0893c84cd302c577b652f235d5e88974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 13:36:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf0235fddb84fb02f540ee16a574d11e860a6980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 14:47:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74dd3c276d4548be8dcdb45dc60c3095\n\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8fb6f7353ace4c5ea9a5227deef41f59 : SUCCESS in 3m 44s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/084ee5ace6cc4c4db06c91d5d9b677d9 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b0952790f14640b8b2cfc78292f341 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/103a34c107bf4a8ba20c008c7cafc1b3 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a61a1667b2247568d7ad2b019ee92d2 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf72f693c8b4eea84175a8b1983ac45 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876c6344538c49c8a853069990550e1b : SUCCESS in 6m 08s\n- osc-upload-image https://zuul.opendev.org/t/openstack/build/55ec300225c94bd5854d498b1d108b7b : SUCCESS in 5m 55s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/0fd73bab4e0647ec99bf539ad27bc94f : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ca617fd9d811feaf35f68808ae88f2ab7c29eebf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 14:47:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aed521d426f45b335e1c69d8b2f6d103c916f75c","revisions":{"af6ea07703ed1c12b86995b97a9618bfa70721ee":{"kind":"REWORK","_number":1,"created":"2023-05-08 10:52:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/882548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/48/882548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/1"}}},"commit":{"parents":[{"commit":"1face4f48b07fbc51824980e989687adaf453f0e","subject":"Blacken openstackclient.network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1face4f48b07fbc51824980e989687adaf453f0e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:32:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:39:03.000000000","tz":60},"subject":"Blacken openstackclient.identity","message":"Blacken openstackclient.identity\n\nBlack used with the \u0027-l 79 -S\u0027 flags.\n\nA future change will ignore this commit in git-blame history by adding a\n\u0027git-blame-ignore-revs\u0027 file.\n\nChange-Id: I2eeade1ce6653be8e9179ecc40105182c5ff5f16\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/af6ea07703ed1c12b86995b97a9618bfa70721ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/af6ea07703ed1c12b86995b97a9618bfa70721ee"}]},"branch":"refs/heads/master"},"aed521d426f45b335e1c69d8b2f6d103c916f75c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-10 09:51:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/882548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/48/882548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/48/882548/2"}}},"commit":{"parents":[{"commit":"6475dc58e8baa199d948c4b2d250c1fb45e0bd30","subject":"Blacken openstackclient.network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/6475dc58e8baa199d948c4b2d250c1fb45e0bd30"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:32:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-10 09:51:30.000000000","tz":60},"subject":"Blacken openstackclient.identity","message":"Blacken openstackclient.identity\n\nBlack used with the \u0027-l 79 -S\u0027 flags.\n\nA future change will ignore this commit in git-blame history by adding a\n\u0027git-blame-ignore-revs\u0027 file.\n\nChange-Id: I2eeade1ce6653be8e9179ecc40105182c5ff5f16\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/aed521d426f45b335e1c69d8b2f6d103c916f75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/aed521d426f45b335e1c69d8b2f6d103c916f75c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
