)]}'
{"id":"openstack%2Fpython-openstackclient~882550","triplet_id":"openstack%2Fpython-openstackclient~master~I6420ec6fd550903b03083b9b1f8391511913c86f","project":"openstack/python-openstackclient","branch":"master","topic":"black","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-10 14:47:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6420ec6fd550903b03083b9b1f8391511913c86f","subject":"Blacken openstackclient.image","status":"MERGED","created":"2023-05-08 10:52:00.000000000","updated":"2023-05-10 14:47:27.000000000","submitted":"2023-05-10 14:47:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882550-black","meta_rev_id":"577ef794bc6205172e10c7c452288d6f802b5af0","_number":882550,"virtual_id_number":882550,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 14:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2023-05-10 13:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2023-05-10 13:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 12:59:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 13:35:32.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"186c273fabca86fe14780114bf69fadcae3cead9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-08 10:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05863d6e43b98d126603430e35d2dd2d834e11b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 12:59:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e41f23f8021044c28367e1adbf5d27bb\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/28c203870e414f8f84e872b69ee3657e : SUCCESS in 4m 09s (non-voting)\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fce5dbdc8e5e41758fd27e5bb84efcbf : SUCCESS in 4m 33s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a4baba92f91f4d0c952eeab86630ee9b : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d16a4750c2d4b00b177d9e38a7b93df : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f8fc18f37d46bcbf6c7c32fad392d2 : FAILURE in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12ed3e81b2f94f4999ad0151a35e071c : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49e9f34a268c47d28ec5e57c24db35f5 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c8a741741e4d089190b7e9e304e28b : SUCCESS in 6m 49s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/bfefff9bfe5e4962a7068ee136dcad5b : SUCCESS in 6m 18s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/131648df87484f8ca999e6073270bbc2 : SUCCESS in 1h 11m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/03b951a60b0c458b933ce94659da7955 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"09e77e3ad768e3fcf57f0dcf6e53f43097fe45fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-10 09:51:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"62800806c06362aa1a956b51943a33e9825babde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:56:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92eb077776dd4d4099f49e29a8824e63\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/110b734575384ee8adba40231c0f683e : SUCCESS in 3m 56s (non-voting)\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6ccb389da9f64f82b2ff8c088356215e : SUCCESS in 4m 14s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1449afb967864718b125a1206c75512f : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3945c98b11194d06bd4180011375d2d3 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a8b6495a6840c887337db223a43aed : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23b6fcc4e314494e8823076bab4eebbc : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aaa93ead2244ba9b91e37e5ada4269f : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23f63a09506e4c32b298616926a492f7 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7d9b52b8a7468a9292b86b788a861d : SUCCESS in 5m 32s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/db460e17b53b49838c820a8504186de9 : SUCCESS in 6m 15s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/6ee43697bdee4ed980753965bbdccb75 : SUCCESS in 52m 32s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4f87db9773d24a55b901a52a7bf3d0cd : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7297fa14419003262d68f3f11f5f31397e940814","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-05-10 13:35:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3c89baa36c2bdf4666e85f4c7928a9d49fdee6f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 13:36:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9a2acd2c53b701aaf07fac968cb22d0002f2826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 14:47:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb248945584747a9b40f43bf7b09db63\n\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ae6a1d76881d4055bd8cf182e7c8fc3d : SUCCESS in 5m 29s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/490afc06e6f94e77857941ce4aeda8a8 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65f33accbc749bfb77ed470ad9c2cf8 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987d8f857e2a4069837bc2018bbc20f1 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c182b844cd4557a1b623b8dc5ae56b : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deb005c7a7c84feba35e7fbce1342593 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24682df536164374ad9ec796e906d162 : SUCCESS in 6m 26s\n- osc-upload-image https://zuul.opendev.org/t/openstack/build/209d235f049b44ab97700e2930501166 : SUCCESS in 6m 12s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/b36aaea30c6544fe999aca6007a130b4 : SUCCESS in 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"577ef794bc6205172e10c7c452288d6f802b5af0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 14:47:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ca43885c26d9dd7903e7015c0a31f3486730eea","revisions":{"53476e1f73d8af172207ac7089fb85fc0221859f":{"kind":"REWORK","_number":1,"created":"2023-05-08 10:52:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/882550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/50/882550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/1"}}},"commit":{"parents":[{"commit":"c51e4ef1bc350905a04690eb019d53e68fe3d633","subject":"Blacken openstackclient.object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c51e4ef1bc350905a04690eb019d53e68fe3d633"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:34:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:39:03.000000000","tz":60},"subject":"Blacken openstackclient.image","message":"Blacken openstackclient.image\n\nBlack used with the \u0027-l 79 -S\u0027 flags.\n\nA future change will ignore this commit in git-blame history by adding a\n\u0027git-blame-ignore-revs\u0027 file.\n\nChange-Id: I6420ec6fd550903b03083b9b1f8391511913c86f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/53476e1f73d8af172207ac7089fb85fc0221859f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/53476e1f73d8af172207ac7089fb85fc0221859f"}]},"branch":"refs/heads/master"},"7ca43885c26d9dd7903e7015c0a31f3486730eea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-10 09:51:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/882550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/50/882550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/50/882550/2"}}},"commit":{"parents":[{"commit":"329296cedd0a64e9f65b080cef749b65256db85b","subject":"Blacken openstackclient.object","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/329296cedd0a64e9f65b080cef749b65256db85b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-08 10:34:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-05-10 09:51:30.000000000","tz":60},"subject":"Blacken openstackclient.image","message":"Blacken openstackclient.image\n\nBlack used with the \u0027-l 79 -S\u0027 flags.\n\nA future change will ignore this commit in git-blame history by adding a\n\u0027git-blame-ignore-revs\u0027 file.\n\nChange-Id: I6420ec6fd550903b03083b9b1f8391511913c86f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7ca43885c26d9dd7903e7015c0a31f3486730eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7ca43885c26d9dd7903e7015c0a31f3486730eea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
