)]}'
{"id":"openstack%2Fpython-openstackclient~970820","triplet_id":"openstack%2Fpython-openstackclient~master~I5758625ea0a7e78913cbd032377fcc9857f8b480","project":"openstack/python-openstackclient","branch":"master","topic":"integrate-manilaclient","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-06-26 02:08:24.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-26 00:22:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"38842":{"account":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"last_update":"2026-06-25 20:30:36.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I5758625ea0a7e78913cbd032377fcc9857f8b480","subject":"share: Import commands (1/n)","status":"MERGED","created":"2025-12-12 11:25:04.000000000","updated":"2026-06-26 02:08:24.000000000","submitted":"2026-06-26 02:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":194,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970820-integrate-manilaclient","meta_rev_id":"67fc80b93ec41866ebcf824e59897995cda53c60","_number":970820,"virtual_id_number":970820,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 02:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-06-25 21:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-06-26 00:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-06-26 00:22:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 12:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 15:15:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-06-25 15:15:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2026-06-25 21:31:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"55641261d0efe98cec8e3787a04c54ecc313420e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-12 11:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abe26a9af120def7f686b9344f23cf27b915317c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:43:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ff2c4fc7ab457989e3d73bb95d6699\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/315ffcd075f5437699e87cbba48135d3 : SUCCESS in 3m 13s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/2c445478f5054c388cd8bf778a8c7d53 : FAILURE in 5m 22s (non-voting)\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/97d65b8ed659423f995437ee11f8af03 : SUCCESS in 5m 02s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/0822f4eb9e414eb0a4b9657ce0e134dc : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e8624effc2f446abd8b45e4f6f5231c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e36a5c3c8f949f7ad707a13c925c2eb : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85d8ae8f90d240b7afea4ba4cc489dd9 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f330cca0d9364f56b365c421f5bc1a76 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a1b1a3ead34d29877e01ff8963a7b9 : SUCCESS in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/faa580e8b4b94cf19dde4132b062633b : SUCCESS in 2m 08s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/258c8223ba934db0af1f7b26c280924c : FAILURE in 3m 41s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/166673940aaf49b58b527a0e07afa6be : SUCCESS in 1h 08m 29s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/da4c82d0ea164409a78d643dfe5f609f : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ddb3a8e3a941da928ab61756f0ce4652d11de2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-10 13:12:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"feb372615b5cefe7937f4d37fdf9cded5d791dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 14:56:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d854095756cd46e8b141374a68cfd017\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c39a3aac84446a3905d25c5867a5a98 : SUCCESS in 3m 35s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/946e06edd91c4d71b98f3aa42ae12d4f : SUCCESS in 9m 00s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/6ac8c7d8944442c2898f6e9b3361729b : SUCCESS in 5m 32s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/eddb3ed4ac5a4d1c829df3c75d5e79b4 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9589bfe10bef42d8ac2353d84eaf5052 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed16e30f34e481a811fe057c1477d2a : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ca7390566c84b7996f692f260bfcb99 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f2d2ed350724015a1615e32aa557e2e : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82034a3f58f1446a8d73e90f19d77daf : FAILURE in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0a8dfdf75d42e88ab5226dced54241 : SUCCESS in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/558db98afce743b3835c005bc6e3b1c7 : SUCCESS in 2m 59s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/2fce223f632148119e2ba14baa250098 : FAILURE in 4m 43s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/2ae1094f50bf417299bcc9ba45b87050 : SUCCESS in 1h 11m 07s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2c103bcb7e3e442bb8afde210af6c250 : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c42dd5898b9d043cc404dbbe100e23c4eec57e60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-15 13:29:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d0edbd6cc67606ec7bcbf64873834b2dc6e2d42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:49:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0253419c4841edb4f190aa198fd0c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59ade0b0ab5a492c93f75d8ed485a240 : SUCCESS in 4m 10s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c85f9ad3f03944fb8780c1adc56944a3 : SUCCESS in 6m 00s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a89ff7e17b8e4076978d23113e3feaa6 : SUCCESS in 6m 53s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e1d98ea7eefb4db0a859d31b31ab3965 : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc8c8b473c3e404990eccafdafd5a7f2 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1bd0addb0f24282a544592a26d948e9 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74d900ad89054788b463ac530c878880 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff5f393ebd3b4966a689d70fa73ca7d2 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4cfb98361dd4e6d8d986bec8f85d2cf : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ff0380d65a438d920e732fd47550d3 : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4c51af6b5284b73b814a15cf598f58e : SUCCESS in 2m 28s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/2de5f16340c0482da03daddae79a8bef : FAILURE in 7m 58s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/87408ba218b048b5b80ac50e8b1206f3 : SUCCESS in 1h 05m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8264a8c8594f4cffb127908dce51d4b3 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7a9ec0374dc4769e8eaef9a06528a844200b1d93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-05-21 09:51:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c1e5f8ecd8eab37b55a9e82f0b82f44c3823597c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 11:15:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f58157a89c044f58ba7ee019d4a6790\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76479c85ab7c438388206fe87f1ca5a4 : SUCCESS in 2m 50s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/843ed0f99ae34d2ea8aa3a27c7b14c2c : SUCCESS in 5m 45s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1db0acd19a4c45d5b1565f5fcac26e09 : SUCCESS in 4m 38s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0104f2e5626b416a8bf39abb78df0948 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d95945c942b14266a7fd9372add25505 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4414cad5b9fe44ecb8e0c5f195f6d181 : FAILURE in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43f214c847f24d30ad1e0a088a2ec4df : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c82bc9c6a73b497ea2cdd065997bde67 : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e2ebc9d882b4d3e8b5d0819a342bf21 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9b118c5fd341c190f00eb3dd8aa788 : SUCCESS in 6m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f35319540ec8448191739b191ea1930f : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c0762c94a694d5aac4a408334710a02 : SUCCESS in 9m 18s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/0bb32b8b342542d4b231868749fc301a : FAILURE in 4m 24s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/d3e0acd37c05428bbcb5ef9797987c84 : SUCCESS in 1h 18m 10s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7bde762dd2464562b8b5088a377625f2 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a68d9f3d1540550c375104ed0ea382ea43a7dabf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-05-22 12:58:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"1220a80acd121ac8f22ad845091d53a76e75d6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 14:26:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8260ceabe1d4bb4a3c53cea26d03208\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21c0708cf460483f96b70a210a2ada61 : SUCCESS in 2m 59s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d73d94e5899545a9a3d19b47539b1340 : SUCCESS in 7m 53s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/092f2dee5d754a7fb021c6837f661cd9 : SUCCESS in 5m 35s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/250ecc4ea34a466b9e8f828ccb8aa432 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/152003a6d0d44cd2ae8619be6a38a45b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe0a4665f144d07af365c81bac1090c : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ca7cc2cc70f4982a27458c87ab2b1b8 : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9510807d948f4ab7b3448a607dd81e44 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8f0073e487146cbb2ea985a01af00da : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b81f2a73aa4909a4aebba3bbe1311d : SUCCESS in 5m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa71e9863d9c462185b16af633309b7f : SUCCESS in 2m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff231389f7040deabae11e38b19756a : SUCCESS in 7m 26s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/1522b4d9ec9e4006973e12111e4cd2a9 : FAILURE in 6m 33s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/e86564a985b54af9b6b846c7cf378bd5 : SUCCESS in 1h 09m 58s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/14afe3fbf1a4435ea0b8b92e6e946418 : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":5},{"id":"d91b425d3d1974dbc65b3b70101ecfc748b9a0d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-04 17:23:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1dc2680b16c992ba609da2a1d9a5d652dce0d290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:23:39.000000000","message":"Patch Set 6: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/736cea4999ce4b17a4996d9bab4955a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb417f13763b4715a500ed86282ea80b : SUCCESS in 3m 22s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e8ce01acaabf48f9b75bfcf4f1f5f749 : SUCCESS in 5m 56s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0dcc46ab31d344daa5e376603dabf6d0 : SUCCESS in 4m 03s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/766aaecceca64695b6bf28a75205a83d : SUCCESS in 9m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7792b5056314956abf408a995844bb8 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42346e80ce2e40be858ef02cb2eeff6f : FAILURE in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f41257d083ad4e6e8bf9b7c6b6c373c4 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99ca4e83648644678b83676f6e7d1947 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/456f1713d598409cbc79b3f45ace4484 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4a8c3eabdb4576911530d5cca1d330 : SUCCESS in 3m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/876f914c968644669f24d751d93457ec : SUCCESS in 1m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f0b633131414a0fbfeba21ab13cc7c4 : SUCCESS in 8m 03s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/7b58013e0c6b41d7af066d2b0b5bf6a0 : FAILURE in 4m 08s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/7aec8da0b36a46a59846044ebd53dca3 : SUCCESS in 43m 36s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/35f635d24f964a86bdcd41ea20a79cf0 : FAILURE in 58m 45s\n\nWarning:\n  Comments left for invalid file openstackclient/share/client.py\n  Comments left for invalid file openstackclient/share/utils.py","accounts_in_message":[],"_revision_number":6},{"id":"abb77c276e6ed327caf10842f18c7f01cdcb4631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-10 09:51:28.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"785674afe9aa700bb2def2f170856f4fc526f9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-10 11:22:57.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"24e76196e3ab5a59e3afea4283e41da8068717f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 12:55:09.000000000","message":"Patch Set 8: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baf12c81d0b949889a21753f70a9b25f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78a4184ecf5947409351cdbfbfd1d948 : SUCCESS in 4m 24s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/403bc2017e654c34b4ea1aa9ca6843c3 : SUCCESS in 5m 58s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/68ceddd30f824ce288ed631e70145710 : SUCCESS in 3m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e8f3cc0712814cc1a6acd0bbb0e5e8ee : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47a1b01d9a174ef9b5b0243d5fa8f778 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/494684c280a945cc87ee0c2c260e6ed7 : FAILURE in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32cb02b76f63419aafaecc8e99f87b33 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64f1c2ac0ca640049beb2f0afdfe5c22 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/532d9b6017144a5d9d9da766261c052d : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c703918bdccf4d2cbaea3cf475a807ef : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3f9da27b1f94cdd94175ca8c50f4e79 : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfacfd12edd142c2ad59acf7c80d1d3b : SUCCESS in 10m 06s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/129cbbf378894c7dae94ea3e4c66aa23 : FAILURE in 4m 32s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c9aea5b4be3a41c9bd02b9a1ad123842 : SUCCESS in 1h 23m 21s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c44812f8a03346c29f11a7e525dcbd9d : SUCCESS in 1h 23m 03s\n\nWarning:\n  Comments left for invalid file openstackclient/share/utils.py","accounts_in_message":[],"_revision_number":8},{"id":"15e5374cef591bc758b111d24696002e876affb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-15 12:12:57.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"6f0376b4dbca2758d31353ef326aabadfc81dfa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 13:20:11.000000000","message":"Patch Set 9: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e78972692714152bf52a44be89f0bd4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/edb2f75c4369409ab976b6c7ba06feef : SUCCESS in 3m 09s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f750c488743d444aa1be19f7c8c474fd : SUCCESS in 6m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4d8d925f51724bdf83d958307e35f085 : SUCCESS in 4m 38s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/30e2a505d49c41c68701fa6cb751fbff : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcda81b5906a447484d96c6eabf7b5ad : POST_FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c3e3c435554e2da91952a3b6b8fad6 : FAILURE in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3321f96137894b279f173c47aca8007f : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76eebaaa382b43b3bfa025076b63d408 : SUCCESS in 5m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2841eb13f3b745c78c953c12ef7281df : POST_FAILURE in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d75f7d3fc44f9c85bd54383add9dd4 : POST_FAILURE in 5m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93f10be3af5c4c889d08979465a95e2c : POST_FAILURE in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69c87b5d1214487aac56394d6a17b836 : SUCCESS in 9m 04s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/bee77c278af14283ad1421dd52c5bd0a : FAILURE in 4m 26s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c05ef45949c942e6ad16c0a86637c902 : POST_FAILURE in 1h 04m 28s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/96c6f1a05d764929abc6c45bdb79a87d : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":9},{"id":"3fb9bdf8fb346283354e0a837f668fa4739d9d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-15 21:08:54.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"e502bbb256bf469b66056cd86caa62bac4b94ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 22:37:35.000000000","message":"Patch Set 10: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/550d2aa3e7e144d784e4b4588b44d82c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1bcd78374b804fac84f464e501700201 : SUCCESS in 2m 50s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0716d2c4f4ee44f8b2c9a5aae51d4cac : SUCCESS in 6m 27s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/707006ddde014f1eb8aebde30379ba16 : SUCCESS in 4m 36s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/725241ef081146be8e589fbfcf507fb3 : SUCCESS in 8m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9221921d7bac4390adabf138343f51de : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a420763b954dfeb53fd7190c477af3 : FAILURE in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d9d30497b8340c98aa8e8f118d599bd : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6c39b3786f04d68a3f497e73d6befa7 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a9a87147b6d4a419416f09ca812050b : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6b6b0fb3314cc784cede225d8492e5 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/220f0464b4f748918bbb0ef7fb17d496 : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c6152b409348a7ad37a45409edc6b6 : SUCCESS in 7m 02s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/d275a5a6a5a349068674ea59b9273224 : FAILURE in 3m 44s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/b370f0eb95d74855afdd4b06132090a9 : SUCCESS in 1h 20m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/29b182ef4a8a4f0f97a32ac5fdd94026 : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":10},{"id":"ab83fdd8e9caf492043e889eb52937059e2788c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-16 09:41:34.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"580f1cb7922e2600545e795dd0fec72eba08968c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 10:45:32.000000000","message":"Patch Set 11: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4c14adbf179450491d6aea129a70df0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d041768351ec4ae8ae2a06c49b65c4a1 : SUCCESS in 4m 00s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ffb8f9772dac4189981efea8c59cd634 : SUCCESS in 7m 28s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fcf3e38a4c4d4f0fb65e444a91b21261 : SUCCESS in 7m 17s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/562cc8e600cd4af6bd838a1ab8088456 : SUCCESS in 9m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff6560e8a99a488fbb19f8ef1d1be0fa : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499def081d9e4dd88fa4e61f46504472 : FAILURE in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8901e7f03ec4503b8929e0e93370498 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2e469e86eed42118edbfc8714ea7c8d : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dee46ed5a65240bbb147e0ad242d5665 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95d25bdceb449b1bfd50e5573e472ad : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c78de7056bb2405a855ee651eb332d67 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/822481e495ca4fc8a3260c6878f4dd31 : SUCCESS in 7m 07s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/cd90bafd91ae471f89bfb2d624b3bcd7 : FAILURE in 7m 34s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/ae02b20ceede43b8bbc6525adccff14f : SUCCESS in 1h 01m 30s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e9ab9ebdc82e4dab8d4e202681409648 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":11},{"id":"634b9ef6d8bd95e75c4403640f54ac934e8ba9ac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-16 15:11:37.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"04849984d8c84f3465818252662c5038773bde3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-16 16:49:40.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"2348ec277edc1eeaa69ec7f2f4039294f7a887e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:25:42.000000000","message":"Patch Set 12: Verified-1\n\n(37 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3168f543814d48599a7952a66288d6e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e7a5cf1706946d389ac85dc9e29abc4 : SUCCESS in 2m 58s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c08dcfcde673405281c0c42cff4b9a0c : SUCCESS in 4m 51s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5668b25ff56e4694829f0c5458772a40 : SUCCESS in 4m 32s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/59c9b03ed6ca42ab93e5ea8dcf5d8e64 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d92046fce547aa82e7304cea7ea70f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ced2f2c5f44bbd940071b2598b3f9f : FAILURE in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a79346f83c7c416eb328902cdae14492 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c173c929b0c74032aa8efa7ff291db94 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b03249c3ff934452bb17275c6e3557df : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cccce7cd3694ea3b143f64bc03faecc : SUCCESS in 3m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a44f30cb705649fa8e430fd7fc6fa41d : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3a99b808f21471091f5a5eeb79b0c9e : SUCCESS in 9m 13s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/6cab588bbbb044698113d613eb337078 : FAILURE in 4m 15s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/6dcab28d52f84a84aa675cf12dcd308e : SUCCESS in 1h 18m 39s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/62499fa0d95040818f09c1b6b77ed669 : SUCCESS in 1h 28m 35s","accounts_in_message":[],"_revision_number":12},{"id":"3c27a53acb1a19e4a725869290e4daa1993fd373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-16 20:58:52.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5109284f0f0d435c21c313853fc60e52d6b9f3ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-16 22:04:57.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"07ed5ca6eb4d91b568b36ea8769ac0b0244aba94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 23:29:41.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08cb19f51438401f92f5ef2b7ca4a3d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c5883670104c4f669540fdae75a50527 : SUCCESS in 2m 45s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/89dae62399e4497bbf8be2fa7cd737dd : SUCCESS in 5m 38s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/843510892c1043129ce36b85e3d99e7b : SUCCESS in 5m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/8fc992e63841435081b355ad2cae859f : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1d76bcf019747029c2595567901ae9d : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9823360696449049a5485d3dab22237 : FAILURE in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b955235a475942f298ed7c6ec70ab8af : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fa5355503ff49ef9a36675ea08699b8 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3ed0c6bcdf442ef8cbb6b4fe3f8e83d : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820c5bb953cb42bf97394a834efac5e7 : SUCCESS in 6m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bdf2cd784c4461083bd447a19d9ef81 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c66e9463edc433b9158431bf268fc07 : FAILURE in 8m 34s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/79bc5affce5d49878c6c5cfd4d83e4b9 : FAILURE in 3m 03s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/cd6d2df0364a4581899da6891b6ac74b : SUCCESS in 43m 46s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b5664c92b68843bb91a1c6c8cbf68dd0 : SUCCESS in 1h 21m 20s\n\nWarning:\n  Comments left for invalid file openstackclient/share/utils.py","accounts_in_message":[],"_revision_number":14},{"id":"11af1f55a490b3a72bfe5229d549942e341970bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-17 10:07:46.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"7ce12465b5075dcd3d593ab6abd934165051f7d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 11:33:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93abebd07473452a9793ff1a40a94f6f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a484d2645f7a4373a463cce9e1680c79 : SUCCESS in 3m 11s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f492973619644fa5b62d371c909921f3 : SUCCESS in 4m 42s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fe94e2fbc23e42978a64714ec77443f2 : SUCCESS in 3m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/cac26203166d4f908faefb6173049014 : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2fad0b16052450da3f34878d17bf651 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4984f4d26fd94ee4991e080373d9e742 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/116517863ec142dc814d2981c90ff6d2 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94a6129476b747799fb70086df7521f7 : SUCCESS in 2m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/849b4cab114a4616975fc8115ed68188 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee558ab89f9489c86d0ad2d011c3a70 : SUCCESS in 5m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7101e781ce774afe8757ce8afc6f5631 : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35156f88d32743beb535f4d6d4812bb5 : SUCCESS in 11m 33s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/9e36a6e4c9114cb1a9d918c96a2bb17b : FAILURE in 4m 26s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c1f1179b76f84312ac08ababcc868218 : SUCCESS in 1h 20m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/83b954e0c0e64f31a9428f15ecaa0ca6 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":15},{"id":"e5f98212b61d438ed2ef41c7a852e107c0226ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-21 11:42:23.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"02f281f95abf90b41376f3d14a63783ddbb4b5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-21 12:11:34.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"f15385a540a015c9ee055b6daca517138f1317c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-21 12:27:23.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"17cfd5affcc73a66ba8a30a58560185989246ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-21 12:36:39.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"6992d9974915f64d323dcb93f11b21fc2c708299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-21 13:23:15.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed77c28ac2d046f588ba99f16896c6d1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/319958107880473f8e2d6b91ebaed8c5 : SUCCESS in 2m 26s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1f2f6abe9eea4bc4adf59d29a1aa67dc : SUCCESS in 3m 25s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c03525b2c8124123bde97335dae67c87 : SUCCESS in 3m 25s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/15ab935b10b24b8794ea9ce6635d4924 : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/753ed933c6ad4238be441c795c55fd33 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842b323c7ac14c0c80c8a1dd7a0c9526 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2b49a85c2544bc38f4f52d00098fb91 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e8658bdbbd842ffb1c7a52cb61ea315 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96b62bac6dcd467b8e19ffe31f0704d4 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4af401fece49d48ee30b6b0b6b45cc : SUCCESS in 4m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/37f366f6f97a4b39b5f77c5b9fb12921 : SUCCESS in 2m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9833346a483b4e38be4e45c5f7a15cde : SUCCESS in 3m 37s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/75f18317eba24dd6ae36492981070098 : FAILURE in 4m 22s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/88b21c045dd944b1821f4dfb7acd587d : SUCCESS in 44m 56s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4c0484a3a12a48feb762daa897791388 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":19},{"id":"4184115cfb2f673a36c11b3409910138cadfff97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-22 11:32:19.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"18696377be6de4e7a8fe85d031a2b982895c82ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:06:12.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b337c685c39f4f72b67f21054cc7c637\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa8fa2f84b114296937eb66a9cb4b8e0 : SUCCESS in 2m 13s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/12510a68e5594cb5b96cff13f5a04076 : SUCCESS in 5m 35s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/66b65d38484b4c459a54990d95db8921 : SUCCESS in 2m 39s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b59337dd0e5d4ed9aa4dc9cf58f762fa : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40734446011c4f68897e79a6d3f5a4b1 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de1966e07384dd0b5c197ed1230d359 : FAILURE in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43cfed81024d47ca82b4a06ae59d4d8c : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32b4f4d24eee428aa99b2f34a5a10a23 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09fa40a0ece148c2a27768ff407a8f53 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50469549a6c74e5bba9662a6d13b015e : SUCCESS in 6m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08c011a46a0f4d7887b6e98ad375bab4 : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30cdfa57d51a40678a17047f6485081e : SUCCESS in 9m 20s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/e2fb42373b994dd8aa83c72313f47d8e : FAILURE in 4m 31s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/6fdfc3b757304539b1854408c00567f0 : SUCCESS in 1h 26m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2960a94b70e34746a8156151aab27f2e : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":20},{"id":"9b57e06d99af97d33d9b36b9ae04f51f6886883f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-23 08:02:56.000000000","message":"Uploaded patch set 21: New patch set was added with same tree, parent tree, and commit message as Patch Set 20.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"1ec132526360dac2710af854a1d21fda42e5d613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 09:29:20.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/136324ced59b4e57b5fb92c155c92071\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6634dd0da80f4a7caa2f499ebb20f759 : SUCCESS in 2m 43s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7ed408c731024fe3b87375a7b56367b8 : SUCCESS in 6m 24s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/851810d5af4640a487ffab1a2456faab : SUCCESS in 4m 33s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/564651074b4a4b0b8ced7bcc6b574a22 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7243b36fb0a743d39809dc85a47595fb : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd68f78ffb54621888a65a44824ca65 : FAILURE in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7049cb0e245429baf1f670c81eb9f43 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc58400683e540b6aef30d7cbc17a7bf : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ccddcd758c4498e8c9fb87f477755e6 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98c653469a84d62b6da8228750eecea : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/889013153c02474ab7ae866f3806450a : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3918ffece9b441ea82eb6dc72352f52 : SUCCESS in 9m 19s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/36ebfd5ee4614f308ec699184c4c42c8 : FAILURE in 4m 59s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/debe14de0d75438383ec16fc12c9f589 : SUCCESS in 58m 40s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5e244be4ff88440da2cf1d5b3faf6260 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":21},{"id":"22db7ef50abd899195bb5aec2a023eba620fd6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-23 10:30:23.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"dcd289c5ed1c4b2c20bea58eedbc437b8ffdf618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 12:02:19.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03051657ca3642ef81f622ce81523a6b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9150baf4704b4a97a8f8901107cd901a : SUCCESS in 3m 03s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5c18e8b33b844e1cab32966497fb82a6 : SUCCESS in 3m 39s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9ee3813e37644ea3a4ffe6c172c0e55a : SUCCESS in 4m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f32e8ef72fc346d18573adf96155892d : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9711de01611c4ae697c14334419cae9a : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2441b60c52747de928ffed83579ffcc : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7866075a26fa4f169ffd075e9387a000 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c542b56b80e47aea66e08fea2feb3b9 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/205ff816f80543fe839f5d8cad611a7a : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff82cc67acba445b893f1811442d4518 : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/31a36f286d5c4b6cbd89e5d905cc6d2c : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07e2a9e59d444385aa290bbc26d696be : SUCCESS in 9m 09s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/999420ab64c74edab10688bd512b9b56 : FAILURE in 4m 44s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/785355cf36514057a9fbde3e2fc37a17 : SUCCESS in 1h 22m 59s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/51273fea54aa4daaa46a55b9822f429d : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":22},{"id":"7fb0a61e032d852c36c71109c9b5c9608721dca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-23 21:21:26.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"f818780e9233c6b15cdd4d063b9d524a973d54e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-23 22:48:41.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"8bf3cedfdb57401156125dfbf5dd4d33fe57a8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 00:22:17.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f4e33b7b84436599789597f8fe6c1a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/48667fe2d6fd4d71abebf086cc5e9bcc : SUCCESS in 1m 57s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/33da8dc4eabe42829dfdb16e24cccf61 : SUCCESS in 6m 28s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b9a7b18d0249461d8f02f1b9604e8b29 : SUCCESS in 2m 37s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0c6cf0b9d6da43e786423d3866f62d04 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c631fa5f164d28bbd01d1a23cfa1b9 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2c682e199840c3a40fc569bdfb5dff : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ea4bc2b453f462089cfd601a0b90cd9 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21014b628fde4c28a1b80c5d5dc0bf6f : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c203841a49034cd69d77aa38b8f0c4dc : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c08e73903347cbbc7bc54b2dce9b04 : SUCCESS in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5415466ef68a4ce6856dfd4f1c656ab2 : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ea636c2fe54f83af16f1a5d15a0d7f : SUCCESS in 4m 42s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/e978f3d7c1c6487ab07de862f113c199 : FAILURE in 4m 19s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/27ae24e8cc19422991decec71d9b4033 : SUCCESS in 1h 26m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8bd54c9aad8c4ed2a20772370f2cd4e3 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":24},{"id":"cb6793f5ae0399d242c77a2a16caf14db965a3c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-24 12:02:46.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"482e255153c96d5e53013d89807d44dc6da1f145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 13:34:33.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffc20ec3dbb64cbf9f92786f503fcfe3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c954e6aeab4241df930125b190731216 : SUCCESS in 3m 19s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/10cbd1d7e62a4d60aafe6d3b5879f4de : SUCCESS in 7m 35s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8ee7406d9a7249cfa186f68294ef144a : SUCCESS in 2m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/73dc78c3f25b45c8aa712cc513c0d8a4 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79d3b8826ab745d6803bcae7d7912cd7 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c185dd67a74d929f84772f7129ae6f : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/706c3816f95b40f09a97191729d0d9c0 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4536623f4e1d441cb0514c6a1475c010 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a115b281554148f798bcac76182968b3 : SUCCESS in 6m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4200ae4128474fe687924ffc5e99c1f4 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b23c26634d544759cb2e8faf1ce3d0d : SUCCESS in 5m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4693f155eb3d45ca87080d42026ffe57 : SUCCESS in 1m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ce18982cd94992884cbe495d5bd206 : SUCCESS in 3m 33s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/c16f02167df441a9b2cf79d670867a4e : FAILURE in 4m 09s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/07b0a42137784e218c9f2a8c1133dac3 : SUCCESS in 1h 23m 15s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/88820d761fbb44bfac9f0ccd3d6d0e04 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":25},{"id":"5bece3cc5605986120a06b2d70215ef2d560661f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 15:15:17.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"40c5fccb954f939432bbb7ebd6b23da72519b9d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-25 19:10:01.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"cf21d18251b398a6a545de92f25ab9daba6d597e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"date":"2026-06-25 19:13:32.000000000","message":"Uploaded patch set 27: New patch set was added with same tree, parent tree, and commit message as Patch Set 26.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":27},{"id":"d2d7faf705f2b7cb8674e74e5f9efab11de5e1ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 19:29:47.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":28},{"id":"61d666ee1e763aa53b02695b97495873b7559ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:30:36.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35e42e9faa5471b9291f93f80069176\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f201bb125534ff18b6e950d51d1b43d : SUCCESS in 2m 18s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d952555f259e4ad6857a1b5c59486ea4 : SUCCESS in 5m 34s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/25aefec9a10840848e312ce428e2202a : SUCCESS in 5m 00s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/256dec48b0e149e3942d336820c5fae9 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/374a69995826447699270c140fe40eaa : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a33aedcba34aaa8037c35ce6b6004a : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3a83ed5eb1f4aeb9727e6c3ecb03185 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d553787fa7694a30ba6987e75ba351b2 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/097ddfcaa8454ebd95a2fdb440f0e52d : SUCCESS in 6m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/14b02a43fd9d4af79f33470ad406160a : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87d8949f237445a82e437f07b44946d : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16631e6b9c094762ad9158a82071a678 : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64b3492558914195959bbfe341689391 : SUCCESS in 4m 59s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/8737fdaaa4bb4e2cbe6c92371f2b1754 : FAILURE in 5m 03s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/332c3ac4a98e4203907bd52825ddb5a8 : SUCCESS in 48m 29s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bf9e129f23714ff5ae0991780925df90 : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":28},{"id":"e700896051174e0cbec0615afb0bc1e24fd0397b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-25 21:08:54.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"d75e8bf99b037f5aa571299ddd8fbb13e872a52c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-25 21:27:13.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"13d7ab26e207827f0c1c6d057fbf60faf60d31ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-25 21:30:09.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"4271cdb1f85f9d477f14d8e51a708c7063f730b9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-25 21:31:47.000000000","message":"Patch Set 30: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":30},{"id":"c66887897f719ee5810da822e6193c3769ed43d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 22:51:50.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63f52c0d6c3240828eaa80e7acee0359\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/45530bb9ce1541bbab5536a28fe1d3cd : SUCCESS in 3m 10s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9de6b09318d34622807009be966dc93b : SUCCESS in 5m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c9d077bc31724b6791d29bfe24042a31 : SUCCESS in 6m 08s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f51387472fbb40998568492bd246562b : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01500c36ff0e43dd845bb1e8d33b408f : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dce960c9154bc48a25f92b0657f70b : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61f5bccc9cfe47a89adfa216078cc8e8 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec13dccdb73f4c3099ec4947e0967855 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8eb9689fc4542aaa29c9eac861ec88f : SUCCESS in 5m 57s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/46b6b987ad81450e9e023a9cfaaa4df5 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e28b87b979466c975641e4397484da : SUCCESS in 5m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6030efcd1064431e8ef44c48248bc236 : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b55a855e10547b0b8706d2cf2f3388d : SUCCESS in 5m 44s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/65aae097d1c849a992e665b99884f50e : FAILURE in 6m 15s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/901f80cef23548589bb83f3d03ad7280 : SUCCESS in 1h 18m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/61b7e87fc30c4a1cadba89fc118ef8ac : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":30},{"id":"491ee23b5c0c8a7b4278e1871693c5fb53744e1f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-26 00:22:44.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":30},{"id":"1c3523dfd93c12df4a0cf4d95f743dd7d2f4ea73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 00:23:27.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"85f038f4e498374907a1b2b59a23ecd8110b3cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 02:08:23.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956282a2868c4852906cc7d501775775\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fad8c5d71c841c3bfc9870f2f92bfe3 : SUCCESS in 2m 30s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d5b0365b719549a4acbcd16fc84bfd26 : SUCCESS in 2m 46s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/78534c9dfee6463a8fad38ace76c6f22 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4f86c0d04e497faf4747f8f1f2927d : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a2139c1ed15409191689665f8338f00 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f55bff67e2a485ea0c2524c4c0c33f6 : SUCCESS in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6cc7c2d1ab2464790e78f53ea384eef : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c0225ebf4b4bbc8b0d1c64409f291b : SUCCESS in 3m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb8cd2851a7e4b3787cf42208aa20023 : SUCCESS in 1m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6adb618ad03d4ccbaad23e2d3af2dabe : SUCCESS in 3m 40s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/6e1c26181c754dbea94f20c5fa75ea16 : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":30},{"id":"67fc80b93ec41866ebcf824e59897995cda53c60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 02:08:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"cc8384de600184b122d1d474da127eeae24055d1","revisions":{"db27780396e9e7aeb36c6824ed4d8a04b7ba24e6":{"kind":"REWORK","_number":1,"created":"2025-12-12 11:25:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/970820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/1"}}},"commit":{"parents":[{"commit":"5f20ffa059d7b9c82ccc52a2eef51b0207ea7152","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5f20ffa059d7b9c82ccc52a2eef51b0207ea7152"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-12 10:32:12.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/db27780396e9e7aeb36c6824ed4d8a04b7ba24e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/db27780396e9e7aeb36c6824ed4d8a04b7ba24e6"}]},"branch":"refs/heads/master"},"193e52b17ff958ce91503a9a35b1ac5c6bb11d30":{"kind":"REWORK","_number":2,"created":"2026-03-10 13:12:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/970820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/2"}}},"commit":{"parents":[{"commit":"1cb15504c02fc7fa7f94ba06de8006b5995a32d2","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1cb15504c02fc7fa7f94ba06de8006b5995a32d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-10 13:12:09.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/193e52b17ff958ce91503a9a35b1ac5c6bb11d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/193e52b17ff958ce91503a9a35b1ac5c6bb11d30"}]},"branch":"refs/heads/master"},"30efd38ff7b559f8c946bea27bca38da73daf47e":{"kind":"REWORK","_number":3,"created":"2026-04-15 13:29:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/970820/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/3"}}},"commit":{"parents":[{"commit":"99a45d77ab1b9dbb59e8b6d7ba0ef5c84d649496","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/99a45d77ab1b9dbb59e8b6d7ba0ef5c84d649496"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-15 13:13:35.000000000","tz":60},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/30efd38ff7b559f8c946bea27bca38da73daf47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/30efd38ff7b559f8c946bea27bca38da73daf47e"}]},"branch":"refs/heads/master"},"7f39cdcd060e00947b3f81424727efd027f004a4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-21 09:51:54.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/4"}}},"commit":{"parents":[{"commit":"efeb55e086242e7efc903a3fcd06e98103fc7833","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/efeb55e086242e7efc903a3fcd06e98103fc7833"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-05-21 09:14:51.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7f39cdcd060e00947b3f81424727efd027f004a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7f39cdcd060e00947b3f81424727efd027f004a4"}]},"branch":"refs/heads/master"},"4e0d930bca38c2c0549b65268676696668797d64":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-22 12:58:25.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/5"}}},"commit":{"parents":[{"commit":"4d2bb3f543125b61790fefa0e3765630bd30035b","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/4d2bb3f543125b61790fefa0e3765630bd30035b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-05-22 10:47:48.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/4e0d930bca38c2c0549b65268676696668797d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/4e0d930bca38c2c0549b65268676696668797d64"}]},"branch":"refs/heads/master"},"f6a1ef008dfcd7d3e0a0a5f39c72de81d76f901b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-04 17:23:12.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/6"}}},"commit":{"parents":[{"commit":"ce3b15b79c9d75035360520379e6d3e958668e7c","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ce3b15b79c9d75035360520379e6d3e958668e7c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-04 17:14:21.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f6a1ef008dfcd7d3e0a0a5f39c72de81d76f901b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f6a1ef008dfcd7d3e0a0a5f39c72de81d76f901b"}]},"branch":"refs/heads/master"},"7b7f3029e81e5a7c4f64901cf3b9e243f8362403":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-10 09:51:28.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/7"}}},"commit":{"parents":[{"commit":"bea9eecf75455bd53f84b0fbf9028d149dee6192","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/bea9eecf75455bd53f84b0fbf9028d149dee6192"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-10 09:50:04.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7b7f3029e81e5a7c4f64901cf3b9e243f8362403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7b7f3029e81e5a7c4f64901cf3b9e243f8362403"}]},"branch":"refs/heads/master"},"d4ec5d378189c25937b4556dcec45ad36fbbb765":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-10 11:22:57.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/8"}}},"commit":{"parents":[{"commit":"57a8edc5ecc891a7e1f7b26f617e9d441d675740","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/57a8edc5ecc891a7e1f7b26f617e9d441d675740"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-10 11:22:49.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/d4ec5d378189c25937b4556dcec45ad36fbbb765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/d4ec5d378189c25937b4556dcec45ad36fbbb765"}]},"branch":"refs/heads/master"},"ef1fc4fce7c7a5e0d6d630cce55298cbbc8aca1a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-15 12:12:57.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/9"}}},"commit":{"parents":[{"commit":"cb15b097b7d41ba8e232c7425fefb5528fb97e17","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/cb15b097b7d41ba8e232c7425fefb5528fb97e17"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-15 12:08:01.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ef1fc4fce7c7a5e0d6d630cce55298cbbc8aca1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ef1fc4fce7c7a5e0d6d630cce55298cbbc8aca1a"}]},"branch":"refs/heads/master"},"4ed05a9af4c54c51f108f73548fdbff6cef6e6ce":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-15 21:08:54.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/10"}}},"commit":{"parents":[{"commit":"c83482b97a71927351dfacdb5ad71c211b4744f0","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c83482b97a71927351dfacdb5ad71c211b4744f0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-15 21:06:26.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/4ed05a9af4c54c51f108f73548fdbff6cef6e6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/4ed05a9af4c54c51f108f73548fdbff6cef6e6ce"}]},"branch":"refs/heads/master"},"f0ab455ed515ca57e213d85c005c1d81a39c0c92":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-16 09:41:34.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/11"}}},"commit":{"parents":[{"commit":"f68f8f5f91d78484858ed877d0cb91ca649a6c2e","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f68f8f5f91d78484858ed877d0cb91ca649a6c2e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-16 09:34:06.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f0ab455ed515ca57e213d85c005c1d81a39c0c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f0ab455ed515ca57e213d85c005c1d81a39c0c92"}]},"branch":"refs/heads/master"},"f664ee63d98e6d7f71c75a8e2324e75b3979bdd4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-16 16:49:40.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/12"}}},"commit":{"parents":[{"commit":"36c9800c71940a96e87a1d4d07e373d5c5d1be66","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/36c9800c71940a96e87a1d4d07e373d5c5d1be66"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-16 16:48:33.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f664ee63d98e6d7f71c75a8e2324e75b3979bdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f664ee63d98e6d7f71c75a8e2324e75b3979bdd4"}]},"branch":"refs/heads/master"},"19493816f7ccbb08cf1b84357843b14d146f43cc":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-06-16 20:58:52.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/13"}}},"commit":{"parents":[{"commit":"2defdb84f1d7eed232f3f56629b0bd56d79ecc94","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2defdb84f1d7eed232f3f56629b0bd56d79ecc94"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-16 20:57:25.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/19493816f7ccbb08cf1b84357843b14d146f43cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/19493816f7ccbb08cf1b84357843b14d146f43cc"}]},"branch":"refs/heads/master"},"2bb553d9674ef72531b20fa4e8cac7b15aaf70db":{"kind":"REWORK","_number":14,"created":"2026-06-16 22:04:57.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/14"}}},"commit":{"parents":[{"commit":"2defdb84f1d7eed232f3f56629b0bd56d79ecc94","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2defdb84f1d7eed232f3f56629b0bd56d79ecc94"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-16 21:58:53.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2bb553d9674ef72531b20fa4e8cac7b15aaf70db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2bb553d9674ef72531b20fa4e8cac7b15aaf70db"}]},"branch":"refs/heads/master"},"01cc8d041a8bc5d38eb4b56b0513596980a2f354":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-06-17 10:07:46.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/15"}}},"commit":{"parents":[{"commit":"49878befe9ee12b3196c533984ba0c140c1dc5ee","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/49878befe9ee12b3196c533984ba0c140c1dc5ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-17 10:07:25.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/01cc8d041a8bc5d38eb4b56b0513596980a2f354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/01cc8d041a8bc5d38eb4b56b0513596980a2f354"}]},"branch":"refs/heads/master"},"49bea10789aaa4fb3d3372d102e1323c941738ff":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-06-21 11:42:23.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/16"}}},"commit":{"parents":[{"commit":"e610b69e41c5a612a939c1024b1db3b3d8e13cc2","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e610b69e41c5a612a939c1024b1db3b3d8e13cc2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-21 11:42:00.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/49bea10789aaa4fb3d3372d102e1323c941738ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/49bea10789aaa4fb3d3372d102e1323c941738ff"}]},"branch":"refs/heads/master"},"2aec7a09da376d715038ca3094fd34421fa92569":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-06-21 12:11:34.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/17"}}},"commit":{"parents":[{"commit":"58c44b0012a4677d4e5a97cfc89bddfeeddaf68c","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/58c44b0012a4677d4e5a97cfc89bddfeeddaf68c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-21 12:11:11.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2aec7a09da376d715038ca3094fd34421fa92569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/2aec7a09da376d715038ca3094fd34421fa92569"}]},"branch":"refs/heads/master"},"9e1ff5d2561ef0cc3bf1e51c3d30aace0cbadebf":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-06-21 12:27:23.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/18"}}},"commit":{"parents":[{"commit":"b37d619d3e85fcae5b32977d4f505139cd79b0bc","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b37d619d3e85fcae5b32977d4f505139cd79b0bc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-21 12:27:04.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/9e1ff5d2561ef0cc3bf1e51c3d30aace0cbadebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/9e1ff5d2561ef0cc3bf1e51c3d30aace0cbadebf"}]},"branch":"refs/heads/master"},"1f3925e63ec06937a8f3d20dcf01e26ab81a0493":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-06-21 12:36:39.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/19"}}},"commit":{"parents":[{"commit":"5f316323e603964be73b96edd06e9f6fa8cc0888","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/5f316323e603964be73b96edd06e9f6fa8cc0888"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-21 12:36:20.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1f3925e63ec06937a8f3d20dcf01e26ab81a0493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/1f3925e63ec06937a8f3d20dcf01e26ab81a0493"}]},"branch":"refs/heads/master"},"0ad3c8aef3a1a6e07ed09bd3f7d281eedebe8730":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-06-22 11:32:19.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/20"}}},"commit":{"parents":[{"commit":"ce61bc3dabe3c12a5579c176f8bc6ac793c808d6","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ce61bc3dabe3c12a5579c176f8bc6ac793c808d6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-22 11:31:46.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/0ad3c8aef3a1a6e07ed09bd3f7d281eedebe8730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/0ad3c8aef3a1a6e07ed09bd3f7d281eedebe8730"}]},"branch":"refs/heads/master"},"270512671f90d7acc91ae96f8985bd8d4f5629f3":{"kind":"NO_CHANGE","_number":21,"created":"2026-06-23 08:02:56.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/21"}}},"commit":{"parents":[{"commit":"3b907b3a53ffb64a3b7cda4000e7ad6611c36783","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/3b907b3a53ffb64a3b7cda4000e7ad6611c36783"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-23 08:00:25.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/270512671f90d7acc91ae96f8985bd8d4f5629f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/270512671f90d7acc91ae96f8985bd8d4f5629f3"}]},"branch":"refs/heads/master"},"c3169793565cfb55f74040a66cd70efd223ce98b":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-06-23 10:30:23.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/22"}}},"commit":{"parents":[{"commit":"d0b42748cdcf64ecb2ce970b737f747ca207d802","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/d0b42748cdcf64ecb2ce970b737f747ca207d802"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-23 10:28:50.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c3169793565cfb55f74040a66cd70efd223ce98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c3169793565cfb55f74040a66cd70efd223ce98b"}]},"branch":"refs/heads/master"},"a0a6653b08a3c619a24209875393fa03dcf4e1a3":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-06-23 21:21:26.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/23"}}},"commit":{"parents":[{"commit":"e677f9ed7ab608086b2cc7b1ac93629fa286f705","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e677f9ed7ab608086b2cc7b1ac93629fa286f705"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-23 21:19:23.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a0a6653b08a3c619a24209875393fa03dcf4e1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a0a6653b08a3c619a24209875393fa03dcf4e1a3"}]},"branch":"refs/heads/master"},"57f772686e4651879d2e657f15aa8d9ef6f96761":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2026-06-23 22:48:41.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/24"}}},"commit":{"parents":[{"commit":"b15ccd2ba708725c220d9cc476a5e311dd1d8c18","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b15ccd2ba708725c220d9cc476a5e311dd1d8c18"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-23 22:46:29.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/57f772686e4651879d2e657f15aa8d9ef6f96761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/57f772686e4651879d2e657f15aa8d9ef6f96761"}]},"branch":"refs/heads/master"},"e6cd676946cb4c0296d87f6200cf27ebbd36d543":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-06-24 12:02:46.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/25"}}},"commit":{"parents":[{"commit":"e57ddcec2463cdeba769353237d7712cf5372aa2","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e57ddcec2463cdeba769353237d7712cf5372aa2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-24 12:01:17.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e6cd676946cb4c0296d87f6200cf27ebbd36d543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e6cd676946cb4c0296d87f6200cf27ebbd36d543"}]},"branch":"refs/heads/master"},"e9e39a03d32315d52052076162383cb0bd0bea60":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-06-25 19:10:01.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/26"}}},"commit":{"parents":[{"commit":"921cc48ad5e81ca523db64f95fceb56330f80574","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/921cc48ad5e81ca523db64f95fceb56330f80574"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-25 19:07:30.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e9e39a03d32315d52052076162383cb0bd0bea60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e9e39a03d32315d52052076162383cb0bd0bea60"}]},"branch":"refs/heads/master"},"c82100fd74be64177ae884646e1fa75116c49845":{"kind":"NO_CHANGE","_number":27,"created":"2026-06-25 19:13:32.000000000","uploader":{"_account_id":38842,"name":"Silvia Wachira","display_name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","username":"Silvia-Wachira"},"ref":"refs/changes/20/970820/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/27"}}},"commit":{"parents":[{"commit":"a811fe998aa12b6c92a403ffa6ced6baf954d032","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a811fe998aa12b6c92a403ffa6ced6baf954d032"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Silvia Wachira","email":"wachirasilvia8@gmail.com","date":"2026-06-25 19:12:15.000000000","tz":0},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c82100fd74be64177ae884646e1fa75116c49845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c82100fd74be64177ae884646e1fa75116c49845"}]},"branch":"refs/heads/master"},"ecda318d1b22f5923d5d36afdd103f57aa9b3f82":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-06-25 19:29:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/970820/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/28"}}},"commit":{"parents":[{"commit":"915001382887f5261971c4051fd15d2c2e0eae2c","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/915001382887f5261971c4051fd15d2c2e0eae2c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 19:29:10.000000000","tz":60},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ecda318d1b22f5923d5d36afdd103f57aa9b3f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/ecda318d1b22f5923d5d36afdd103f57aa9b3f82"}]},"branch":"refs/heads/master"},"8aed63421ca2e017dbc96e6005e55e34fd900178":{"kind":"REWORK","_number":29,"created":"2026-06-25 21:27:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/970820/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/29"}}},"commit":{"parents":[{"commit":"915001382887f5261971c4051fd15d2c2e0eae2c","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/915001382887f5261971c4051fd15d2c2e0eae2c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-25 21:26:23.000000000","tz":-420},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8aed63421ca2e017dbc96e6005e55e34fd900178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/8aed63421ca2e017dbc96e6005e55e34fd900178"}]},"branch":"refs/heads/master"},"cc8384de600184b122d1d474da127eeae24055d1":{"kind":"REWORK","_number":30,"created":"2026-06-25 21:30:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/970820/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/20/970820/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/20/970820/30"}}},"commit":{"parents":[{"commit":"915001382887f5261971c4051fd15d2c2e0eae2c","subject":"share: Import utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/915001382887f5261971c4051fd15d2c2e0eae2c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-28 16:59:23.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-06-25 21:29:28.000000000","tz":-420},"subject":"share: Import commands (1/n)","message":"share: Import commands (1/n)\n\nImport the following commands:\n\n* share lock *\n* share message *\n* share security service *\n* share service *\n\nChange-Id: I5758625ea0a7e78913cbd032377fcc9857f8b480\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/cc8384de600184b122d1d474da127eeae24055d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/cc8384de600184b122d1d474da127eeae24055d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
