)]}'
{"id":"openstack%2Fpython-openstackclient~995991","triplet_id":"openstack%2Fpython-openstackclient~master~I8060005332d2c8af91986b86cef4dfe42dd7d048","project":"openstack/python-openstackclient","branch":"master","topic":"migrate_stadium_osc","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-09-02 23:01:14.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-10 14:19:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-10 16:09:35.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-02 09:07:38.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I8060005332d2c8af91986b86cef4dfe42dd7d048","subject":"Network: Add onboard_network_subnets","status":"MERGED","created":"2026-07-03 17:38:19.000000000","updated":"2026-09-10 16:11:55.000000000","submitted":"2026-09-10 16:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995991-migrate_stadium_osc","meta_rev_id":"95589c9c40fafd7caced5d600072b2c5e528e074","_number":995991,"virtual_id_number":995991,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-01 13:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-08 20:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-10 14:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-02 09:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-10 14:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-03 19:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 09:59:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-11 13:30:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-09-01 13:36:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-01 15:39:32.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2026-09-01 18:47:21.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2026-09-02 09:07:54.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-02 23:01:14.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"f9f8d7a043a1bf9b0457891551d04a151aa6af04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-03 17:38:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d5e65b730c12b5ced7a33eeb99fc3243f8ccdf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 19:03:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c65cb11fa3248acb00bf7e998c7925c\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/91e48ae5c0ff4b30a117a28a7b816ec5 : SUCCESS in 5m 54s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f5ab17414134486f87f5ce69826d974d : SUCCESS in 4m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/db4bdb83f24649e6ae68256511dd0c0d : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c560da661024544bbd25e468d6bb055 : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e087a038bd48469c9456fb8c418cdee4 : FAILURE in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17cfc6f42c294fd288ee4996a15dec2c : FAILURE in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d261f0160894d0888a4f46138ff4028 : FAILURE in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9f0617994d74df78683dc0a7bbdf037 : FAILURE in 6m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6691cfb50a99476b812a29c6a35d6ec3 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/022aec38f6b94e2a8503391e06156330 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e64d22bf3e34400ab04a23bf2d0b96aa : SUCCESS in 9m 08s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/4a3cfc64f7f74cba8a590168401e559a : FAILURE in 4m 32s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/1e4fed7708a44e7aabca6bd9ece15b81 : SUCCESS in 1h 18m 57s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/07e0a83d4da148a5ba84cf2d9b7d7ec0 : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"420b3fc47b3c93f16bded60ec53f3dabdb0c7cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-07 15:33:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"93ae8f9de99ade1ad036bd9d32209625f7a6003b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 16:59:50.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b599613d721545e19e7dc3f6d7880682\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3288aeae1d946a0a71fb3b8168ea789 : SUCCESS in 5m 02s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3bd4f85126734ea881b57fadfed6f1e8 : SUCCESS in 7m 49s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1a53571f6e554f3484edf397ce60363e : SUCCESS in 4m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/79e0e41b9d4945198a9137505b7671ce : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7eed538e6ba4f1aa8c9e5cb0e756301 : FAILURE in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676467bdc7474913807eb14cee9fbf43 : FAILURE in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2886e3949b164181ab9bec59766a09b1 : FAILURE in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e31a00fdc6074515a53190d15aead0e8 : FAILURE in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96259cc5e54048ea95fb42e843383b2a : FAILURE in 7m 44s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ad52c917f7444736b46a31784421b799 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7abc21259b4f448ee6fb5915ccf782 : SUCCESS in 7m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f18051796f734c8884e5ab8c79481174 : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/031c4483ebaf4b2bad8eb6e1658ba2ec : SUCCESS in 9m 30s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/b3061fc1524b48d8b5ab91c41f174ae2 : FAILURE in 7m 10s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/23b81ea9c8464bc78839574f90fcb340 : FAILURE in 1h 16m 17s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6a60b069269e4b8bb128cf9b1fdb9f80 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":2},{"id":"695018c3e0f4e0a4c6224ea1de00b83db4e365d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 09:59:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c5a74a4e36476c7734ce527813fe4f97b13c49b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-12 11:26:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d0e6a1c1004cafcc18738bdc02f49b7371bb1b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-12 12:12:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"caa18bb390cf19b658d7c42f13defcdcbfa51e62","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-12 12:13:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"add2be4a162e909cd087edfbcafc6de2c536e1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 13:30:12.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abed0de4edeb4a4ab7d8e603ac2f7128\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad909584c10a4f1e89579a07fcba7607 : SUCCESS in 2m 44s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/20ca0dd260d947ceba0bbfe42a942209 : SUCCESS in 6m 05s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c8ba967166b6414a89fbc8108d2c0164 : SUCCESS in 4m 22s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/8b4fb7d2434e4af494069cf4c053ca7a : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12069241d34740cfa6ade5d4201d5ff6 : FAILURE in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb28f7eb0fd450982e04f727201b9aa : FAILURE in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94a272c8ba9c4ee9909bc86c9ed6b9cb : FAILURE in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dacb9b7e1634f6eae2ec3d2258cb009 : FAILURE in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db631e10f40a4e3db92e7c7cf29e32ff : FAILURE in 7m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/40a7d16af6c645e48b89bf597b36742c : FAILURE in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2892cd0193474d80fc709e7a722954 : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/537a7c0a7ae24d988d106a1abd190633 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d477cf96c5774402b5f452bafd0d8774 : SUCCESS in 9m 05s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a23c088b102546fdb8b014783d59a69b : FAILURE in 5m 00s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/94f00108bf9a404aa3086ad7f8839f98 : SUCCESS in 48m 41s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a81248652f2f40d7bb54689c7ca2ae13 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":3},{"id":"26c5c83e9fd1a0889a0d5ea4bb5e456e10871fe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-01 08:43:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"908e6b0d166fb665e405aef57d613057157dbeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:30:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e5f730d827348b295234f540935b442\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3572518ffac944189c7169981da1f740 : SUCCESS in 3m 17s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7504cd1400be4af8af463adaa1d2d53b : SUCCESS in 4m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7fda7c4b4d2048b397fb34d1020bb86c : SUCCESS in 4m 22s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ad46ae49a8de46a68f1d7657ff3cab71 : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abe74b8d50224b1ca43383035bb06737 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0095afb835246b49164cfb3d2074e00 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23234a95bf914a618ab7782176457ded : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2f12f0c34904f2a8b91bb5faeb32813 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69f9b56f9df84853b52cb25988156b11 : SUCCESS in 9m 24s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bba854b07c7c40598b2361957d701500 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2b456a75274c5cb41dd786b6073e82 : SUCCESS in 4m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70c93c18c566464ab4567fed889c059f : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/136eae5469cd489c86c3e2e1d06ca350 : SUCCESS in 9m 12s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/59ec96f181be4cddb1addebce540eb53 : FAILURE in 4m 15s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/1ad02478ba7942f5b7cc56eb27667c64 : SUCCESS in 45m 57s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5d9655bf7da04df096b6e50b44228665 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":4},{"id":"aa769368fa047ff5f944f6c71b83c0b7c3345135","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-01 13:36:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dcc06116a13455376c4baf9e049f3af422d8b1b3","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-01 15:39:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be2a390a0e666f7545b3b11a5f23e63a5f60fc8d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-01 16:59:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9dcfca598e95029f307232faa71d65213fa20ba2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-01 16:59:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b939d1cd4fea970e9be323d880c970c0a238526d","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-01 18:47:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f192405e861bd16a405ff4ba11661eb4e13eba35","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-01 19:46:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a9f3d079d2a08857e71f563574f6d7f64bfe2b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-02 08:54:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"13628f427414649b8a988f5ed44e3a7fba7ec674","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-02 09:07:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16316a6fd5f460e9eef582f4c98a6f488a484b9f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-02 09:07:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c66e04dbefdaab25d26d8b9b81953b324d0d2167","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-02 23:01:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f0799efb01297a7690d3c352838e8192b4f8dc28","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-08 20:03:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f9f59493564d5d6c35ecba24702f3cac8c38dfd8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-10 14:19:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"860935d0f29e6e08936433806bdd0d9d083672bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:20:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ccdd8fd4188eb9918da773c9842ae711c88f8b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:09:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f394289c60814c38a2aef38fa1f8e84f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6ff2850c3b2403dae317a0b66a86d9f : SUCCESS in 4m 04s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/00e11ffdbb7a4e9a81d22136910e4165 : SUCCESS in 7m 55s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0296116f2076459298d9277e206b43a6 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417ae2965d984031b11e7512770f2662 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4bc504351224316b28057c62f0a53b9 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e51774178d99421a84dd0b2a83cc8630 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b8b40202de64d9e85541d082b3b0002 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08429eaa1b04837a57ea86dc9ae882e : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8bf8bc2974a34e18939a616040d7db5d : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45b45b1949cd43e7a7a337626e0e3196 : SUCCESS in 4m 53s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/1a9940e5256a4d2ebcf2c2d81ae50538 : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":4},{"id":"74a7e43dec4da517fd684db7b6af1afa1883c2a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:09:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"95589c9c40fafd7caced5d600072b2c5e528e074","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:11:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e32ffb237b740379eee44ac41fdb8e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d7e04064ba4c0eaaa4e5c57de4ff70 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90b3e1fcb23f4b2f8904f153b7922941 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7cf74120cdc7be8d4ab4963d37aacbb86edac612","revisions":{"a03eae6a50affc31ccc8678e5e4973e32d82c705":{"kind":"REWORK","_number":1,"created":"2026-07-03 17:38:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/995991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/91/995991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/1"}}},"commit":{"parents":[{"commit":"caec5c3f54aa5019a3a957e9fc3679b1584d4814","subject":"share: Add support for AZs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/caec5c3f54aa5019a3a957e9fc3679b1584d4814"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-03 17:36:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-03 17:37:57.000000000","tz":120},"subject":"WIP:Network: Add onboard_network_subnets","message":"WIP:Network: Add onboard_network_subnets\n\nDepends-On: https://review.opendev.org/995969\nAssisted-By: claude-opus-4.8\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nChange-Id: I8060005332d2c8af91986b86cef4dfe42dd7d048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a03eae6a50affc31ccc8678e5e4973e32d82c705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/a03eae6a50affc31ccc8678e5e4973e32d82c705"}]},"branch":"refs/heads/master"},"885128d6fcf192f3529e749c792fceedad99008a":{"kind":"REWORK","_number":2,"created":"2026-07-07 15:33:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/995991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/91/995991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/2"}}},"commit":{"parents":[{"commit":"caec5c3f54aa5019a3a957e9fc3679b1584d4814","subject":"share: Add support for AZs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/caec5c3f54aa5019a3a957e9fc3679b1584d4814"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-03 17:36:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-07 15:32:05.000000000","tz":120},"subject":"Network: Add onboard_network_subnets","message":"Network: Add onboard_network_subnets\n\nDepends-On: https://review.opendev.org/995969\nAssisted-By: claude-opus-4.8\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nChange-Id: I8060005332d2c8af91986b86cef4dfe42dd7d048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/885128d6fcf192f3529e749c792fceedad99008a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/885128d6fcf192f3529e749c792fceedad99008a"}]},"branch":"refs/heads/master"},"b3987e940cd443a84ef10e86104edbc9ee774f6a":{"kind":"REWORK","_number":3,"created":"2026-08-12 12:12:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/995991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/91/995991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/3"}}},"commit":{"parents":[{"commit":"01668c8b1ce3bd5ad633d2ff849755908c30ab7e","subject":"Merge \"Add support for passing image size to Glance API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/01668c8b1ce3bd5ad633d2ff849755908c30ab7e"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-03 17:36:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-12 12:09:04.000000000","tz":120},"subject":"Network: Add onboard_network_subnets","message":"Network: Add onboard_network_subnets\n\nDepends-On: https://review.opendev.org/995969\nAssisted-By: claude-opus-4.8\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nChange-Id: I8060005332d2c8af91986b86cef4dfe42dd7d048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b3987e940cd443a84ef10e86104edbc9ee774f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/b3987e940cd443a84ef10e86104edbc9ee774f6a"}]},"branch":"refs/heads/master"},"7cf74120cdc7be8d4ab4963d37aacbb86edac612":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-01 08:43:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/995991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/91/995991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/91/995991/4"}}},"commit":{"parents":[{"commit":"926da721a80fba06a3dd39be06f0505305d29d75","subject":"Merge \"Add a \"cloud cache delete\" command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/926da721a80fba06a3dd39be06f0505305d29d75"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-03 17:36:52.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-09-01 08:43:28.000000000","tz":0},"subject":"Network: Add onboard_network_subnets","message":"Network: Add onboard_network_subnets\n\nDepends-On: https://review.opendev.org/995969\nAssisted-By: claude-opus-4.8\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nChange-Id: I8060005332d2c8af91986b86cef4dfe42dd7d048\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7cf74120cdc7be8d4ab4963d37aacbb86edac612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/7cf74120cdc7be8d4ab4963d37aacbb86edac612"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"01668c8b1ce3bd5ad633d2ff849755908c30ab7e","ours":"b3987e940cd443a84ef10e86104edbc9ee774f6a","theirs":"926da721a80fba06a3dd39be06f0505305d29d75","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
