)]}'
{"id":"openstack%2Fpython-openstackclient~996301","triplet_id":"openstack%2Fpython-openstackclient~master~Ida1608a0d8753e555206da56312a85536e50ba5b","project":"openstack/python-openstackclient","branch":"master","attention_set":{},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-08-05 11:28:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida1608a0d8753e555206da56312a85536e50ba5b","subject":"Allow removing deleted projects from flavor access list","status":"MERGED","created":"2026-07-07 14:57:42.000000000","updated":"2026-08-05 11:31:25.000000000","submitted":"2026-08-05 11:28:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996301","meta_rev_id":"1f89f31e2d274c78ad724cb645f66eeaab317def","_number":996301,"virtual_id_number":996301,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-05 11:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-05 09:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-05 09:37:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 16:47:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 09:37:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a010ffcc008204fae0ffb041986e3bfdf94127b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-07-07 14:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0036f45cfe9438937eb5e6e110c26d3748eeefd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 16:47:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e311481dc5459c953567b1c79a5dac\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/56295430b14c4fc49afc0c3534927e1f : SUCCESS in 7m 37s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fd022312e742486cb9f7739b878d53b5 : SUCCESS in 5m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f4646d5e32964c5bb1da6c8fb799dea0 : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94d8592ce56c4b8a96c67e973b18a6ec : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d21b991aa3ea4d2a8b8d6ebb2fe150cd : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5726b6953804334afd9164dc8afa63b : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4f8d378a11146b0afa384b2aa65191b : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0b0b16831bf45448ce77f143383f4b1 : SUCCESS in 6m 43s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/941d77a7fa8c496c9640540d631c8c7e : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e4b8e13904465a91965039865c634d : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61996eab03c244488739dd2e0814ad78 : SUCCESS in 7m 01s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/36f8aea4378c41289b901d202ab341d8 : FAILURE in 5m 37s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/6725dcbe72674108b57662992810a050 : SUCCESS in 1h 21m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f2100a62344a4c2b9308e982f064aed1 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"717d58e1d8b5a003ae63cdb10ee55327b91c9129","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 09:37:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09829cfcaa6ca0e3ab3e7d67630405b3ea444a9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:38:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f1411b72f125be0f3fd0d23330444d289aeec8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 11:28:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21db1e447f4e408f8fd6e2b83275f5c7\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a80cac62b0c442ae8bad14b204356a8e : SUCCESS in 4m 51s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/835850faa1334a1ba8715588117bc3a5 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f699b7d091e2401591da595a30160b9a : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa1c163b2e714f8da1e2e61f61c071a8 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7de14bbda3044ed9d8f28821e14d03d : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c8b72af23eb4b5c966b2d7aba797817 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59b2e2a80654939a6b6c8ba18e2a9ac : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e210b7a96f0f48b4942d0021d293b2ca : SUCCESS in 8m 32s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/c622433f252747d1a92d29d5632af99b : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"81ec0533c623fbd07c4f9376ff69caed12873ab4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 11:28:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f89f31e2d274c78ad724cb645f66eeaab317def","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 11:31:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4196726bb6634decba0a8cd772128064\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55bb235fcf5d4341978bb5622218a7f8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12ba8638936db9a8dadfd336209286ef35ea5e2c","revisions":{"12ba8638936db9a8dadfd336209286ef35ea5e2c":{"kind":"REWORK","_number":1,"created":"2026-07-07 14:57:42.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/01/996301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/01/996301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/01/996301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/01/996301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/01/996301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/01/996301/1"}}},"commit":{"parents":[{"commit":"f6c9939ea72cda93b3ca89813e2fc91e5d8bc705","subject":"Add reproducer for bug #2099702: flavor unset --project fails for deleted projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/f6c9939ea72cda93b3ca89813e2fc91e5d8bc705"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-07-07 14:53:28.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-07-07 14:57:07.000000000","tz":120},"subject":"Allow removing deleted projects from flavor access list","message":"Allow removing deleted projects from flavor access list\n\nWhen a project has been deleted from Keystone, `flavor unset --project`\nfails because find_project_id_sdk() validates the project existence\nagainst Keystone before calling Nova\u0027s removeTenantAccess API.\n\nPass validate_actor_existence\u003dFalse so that when the project is not\nfound in Keystone, the raw project ID is passed through to Nova, which\nalready handles deleted projects gracefully since the fix for\nbug #1980845.\n\nCloses-Bug: #2099702\n\nAssisted-By: Claude-Code opus 4.6\nChange-Id: Ida1608a0d8753e555206da56312a85536e50ba5b\nSigned-off-by: René Ribaud \u003crene.ribaud@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/12ba8638936db9a8dadfd336209286ef35ea5e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/12ba8638936db9a8dadfd336209286ef35ea5e2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
