)]}'
{"id":"openstack%2Fpython-openstackclient~997058","triplet_id":"openstack%2Fpython-openstackclient~master~If65ca0cf09e6d41a4eec602556e23fc489802d0b","project":"openstack/python-openstackclient","branch":"master","topic":"remove-v2-service-set-test","attention_set":{},"removed_from_attention_set":{"39062":{"account":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"last_update":"2026-08-24 23:28:51.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-24 21:51:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"If65ca0cf09e6d41a4eec602556e23fc489802d0b","subject":"tests: Remove \u0027volume service set\u0027 functional test","status":"MERGED","created":"2026-07-13 14:17:45.000000000","updated":"2026-08-24 23:31:21.000000000","submitted":"2026-08-24 23:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997058-remove-v2-service-set-test","meta_rev_id":"b014e2820029f050bbceb3beafc7adf0767ead3b","_number":997058,"virtual_id_number":997058,"owner":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 23:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-24 21:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-24 21:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-13 14:18:43.000000000","updated_by":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-07-13 15:37:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84ea09248881ab618a38e99a8c4676cfe1f57a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"date":"2026-07-13 14:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c0cefa6897138e6604d4e71b11841d83e64a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 15:37:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17cb2275324a479dbd442c71d02c9ba8\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a2a80f83d5d54cf89a05ae970b7a11c0 : SUCCESS in 4m 57s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/27055697c7ea422eb098dd8277087425 : SUCCESS in 9m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e9a9abf93f440d699cd1855c2e1600a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ff5991852046d7b588151d31bd1262 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b232a8fdf9749bab48873e368ae2202 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1705b3b97d174ba59237075b59f08a84 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14355874e10548f0914df72df0e9ed76 : SUCCESS in 3m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4a061ba64c37477ab8f0ceadb1eec285 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d151943de3f44ae84d963b94f00f84a : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2952284a6a454bb81c98da828fdeaa : SUCCESS in 7m 39s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/2e98c441511d48c88585a39da132cdb4 : FAILURE in 3m 36s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/1aaf4f20b9d44dfa89240a24b6b83e21 : SUCCESS in 45m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3a615eb2ed0e4802bb47f689a96f86e2 : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"20d4b4c67841640db25284f6a27b3ab6affcad1f","author":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"date":"2026-08-09 02:31:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"583ee2cbaba319e4a73d0c642c281de0a6e611ab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 21:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69cbf6a5c1c6934f7bd58b3f1809589ac1ffd2dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:51:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0436929df282d64b615afbca58da6957fa678c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:28:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e04e2086fd44100ac58fac7415dc804\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bf4a60131f7946e8a47cc8de007ee4df : SUCCESS in 4m 49s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b75e55acc34c4c4bb5aa77ab23d16efe : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0388133da8064562bbecee463b30545a : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e8b925666324fafaedf52f3a18c99fc : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9109c3f570e74ea3a1f49c260026bad8 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57623015951249d4afdf663ccc0bb2c4 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7709bfde86416a8a2080611a36eff6 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dd511446a75466780d5ba2f9ed9f57b : SUCCESS in 10m 09s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/bd5353d1f0984aea9bf3dc4cfe806699 : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4336b728f37c3c6b827ad010efcfc9856d093d68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:28:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b014e2820029f050bbceb3beafc7adf0767ead3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca45d50320094a05ac0ba1b082e2c143\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b29f240d1641e2aba0c823f10ed536 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8efcc8e8980c35364c2ae6b02393c05f6dea187","revisions":{"c8efcc8e8980c35364c2ae6b02393c05f6dea187":{"kind":"REWORK","_number":1,"created":"2026-07-13 14:17:45.000000000","uploader":{"_account_id":39062,"name":"Park Kyumin","display_name":"Park Kyumin","email":"gmpark007@gmail.com","username":"FrontHeadlock"},"ref":"refs/changes/58/997058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/58/997058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/58/997058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/58/997058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/58/997058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/58/997058/1"}}},"commit":{"parents":[{"commit":"caec5c3f54aa5019a3a957e9fc3679b1584d4814","subject":"share: Add support for AZs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/caec5c3f54aa5019a3a957e9fc3679b1584d4814"}]}],"author":{"name":"Park Kyumin","email":"gmpark007@gmail.com","date":"2026-07-13 14:10:07.000000000","tz":0},"committer":{"name":"Park Kyumin","email":"gmpark007@gmail.com","date":"2026-07-13 14:10:07.000000000","tz":0},"subject":"tests: Remove \u0027volume service set\u0027 functional test","message":"tests: Remove \u0027volume service set\u0027 functional test\n\nThis test disables a cinder and never re-enables it,\nso it can race with other tests running in parallel,\nwhich is the concern raised in review [1].\nSince v2 API was removed in Xena,\nIt would be better to delete than to modify it\n\n[1] https://review.opendev.org/c/openstack/python-openstackclient/+/957586\n\nChange-Id: If65ca0cf09e6d41a4eec602556e23fc489802d0b\nSigned-off-by: Park Kyumin \u003cgmpark007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c8efcc8e8980c35364c2ae6b02393c05f6dea187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/c8efcc8e8980c35364c2ae6b02393c05f6dea187"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
