)]}'
{"id":"openstack%2Fpython-openstackclient~998212","triplet_id":"openstack%2Fpython-openstackclient~master~I0cead57c3db621d25e257e4bc572aff544f9cbb7","project":"openstack/python-openstackclient","branch":"master","topic":"bug/2161353","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-10 10:38:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-08-19 16:48:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0cead57c3db621d25e257e4bc572aff544f9cbb7","subject":"Add leak-routes related options for subnets","status":"MERGED","created":"2026-07-21 19:11:31.000000000","updated":"2026-08-19 16:49:06.000000000","submitted":"2026-08-19 16:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998212-bug/2161353","meta_rev_id":"bbe9e238d3a85d922c7fc106d72bf609f4cb2327","_number":998212,"virtual_id_number":998212,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 16:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-10 10:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-08-19 14:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-19 15:37:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 20:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 09:32:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-19 10:29:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"ebc7c623e9f542620b1c775a9b231ded043cee13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-21 19:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e1dc19687da7fbaed4a948b0aae1e6c3319fe51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 20:30:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a8dfb363a746daa90fbd4c9f6fd6c9\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a1eed68f43294c54ac754dcb71743ad4 : SUCCESS in 3m 05s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4dc0fff6e6fb4e12bd5a59db358c094f : SUCCESS in 2m 46s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b6c32120cb204f298f2a6d8031ef3ca3 : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b6dcff529ae47f389ea230ac974b12f : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1dc3a04b52491d9dce4a691af4e538 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ea94ef09dff412ba22abd1eb2c2654f : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37e1dc0e668e41c982dd543b7eededfb : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1dbee7154dc407988207d027fa26f1c : SUCCESS in 3m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7264d7a98ae7467080c9c70b5dc609ff : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1414d65ecf24427aa9d4bbb7e2785dc1 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eee59b8dace4dbd8d955d2b36bc22b3 : SUCCESS in 4m 54s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/97bec8fcb8564495b1a3f73829565e71 : FAILURE in 4m 25s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/190d9380347b48a4907e9675afdf0131 : SUCCESS in 1h 11m 36s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/be8ccda10e834208b3dbe04e59e43ccc : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":1},{"id":"89085424c71948e22b79e68fb5b34cfac6df90e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-23 19:51:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d318a88baa7a4c07d4c18e6c5c858e3719db9f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 21:08:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9893937c02a44dca7586fc72ede5de3\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5c91789c89f04293ba7017f2189cb3bd : SUCCESS in 5m 28s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9fe2bf129df64e7db77a38d37c4c3c4c : POST_FAILURE in 4m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/98c193995c0a4c9199833926caf5d061 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53befce01ddf4f4182afc4f16a2f3314 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c679b39188345a19f7e77087e863c37 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d94b1c9065ca4585b0e1ddd2b1e35c2a : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c09c573d488443b9bf4443e0ab4793ee : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df875f0eb81144c79f6983a6c6bc3f0a : SUCCESS in 6m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3a257cb697de47aeb9868c91ae7b107e : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9932180fa645278f0a79ce8c672dfc : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/276f4c2271394295a4efffa51a3ce343 : SUCCESS in 5m 23s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/d7c6e221bbbc484786bda71983c93bd5 : FAILURE in 4m 40s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/043a7b1cc8bc4808aadfba687924db3c : SUCCESS in 1h 11m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/52c41b8387a54ed8a4feb359b0e49976 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":2},{"id":"fffb619e882506d0e3b08188abb40acee092dbf0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-24 13:09:36.000000000","message":"Topic set to bug/2161353","accounts_in_message":[],"_revision_number":2},{"id":"295b727ab4e3811f6b42cea4abde08119c24ba7a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-30 14:22:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"af25a98d94dbebebb797aebfb9c613076b5b502b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-30 14:22:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c3c305995623f41910a56c9e7f1b4b2ce7eb59a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 15:47:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e0f6118e4d40f1b3f0df0003179341\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1c49d71f2ce64e2cbd0391f7b47a64bd : SUCCESS in 3m 00s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cd9030bebec041e8913a81564ce7f67e : SUCCESS in 4m 52s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/56764c6e5f3a41b985aa03392e10ce9d : SUCCESS in 9m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5162c2149ece473a8dcfa81e0f268cb4 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/420aa83764584775a3ccbb67e16817f8 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4912e20c854348ce821823f37a6f2c98 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aff314a19172461e81b3465d8f953520 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85f0e50b36b544f1a34c57639a57d88f : SUCCESS in 9m 21s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/cdcb08ce6f044a4ca5735a1000d3b9bf : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65faa8469cb43dfb910a3b7d2fb6dc1 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a021eee30b52478ebf6a8ab56ebf0822 : SUCCESS in 7m 33s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a44c2acc2444412189d7b639a3b7e30f : FAILURE in 3m 50s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/b3bbcab3f404497997d8e16675d0efb4 : SUCCESS in 1h 09m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b70bf0c2faa14bccb46e25193190d7cc : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8679032bd9b8bf23ce3b3e940e9773a9761dfd32","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 09:32:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5d4fb59423919efd741548528eaae5ed9a07e5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-08-06 19:42:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"653b8dc037a821e73af1388355881e417cf2790c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-08-06 19:43:39.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb6c81ca96ef53b84d23c14c36fc5654d6661a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 20:58:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab85a39efbf4035bf9e2ba108a4cc8b\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d75d7825947f4c5f9beeb5c7aea8b157 : SUCCESS in 5m 16s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/af1baf37f2ca4ce5b4653d2500db60e8 : SUCCESS in 4m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0ce3130af36c4f0e99fb16cb70b8fe07 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be72eddf6ac7458abe5c2873757445b8 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7997b8daf7a743c7a9663fee85b54e28 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b7e5c28807d4a8d9989a57ffca1724e : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f82904eb776b4617a432c0cb437cce83 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a22c66cad2e74274ada1e5ffb9d28d05 : SUCCESS in 6m 57s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/43d7cedbb7b947d3862a086cc6276ed1 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb20d82f6ba94e2d8c63eda23cde18bf : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8fde6fcdd3b4c67b6655084c65a6fb7 : SUCCESS in 8m 46s\n- osc-build-image https://zuul.opendev.org/t/openstack/build/a742d38a67674dc58e9fde3b85d71c93 : FAILURE in 4m 21s (non-voting)\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/449fa67d25154f2ca50e0095e1911296 : SUCCESS in 52m 01s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1791233ff96f4c6ab7165209b2a770a4 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":3},{"id":"2a048b80d4f079699cb2d50f0fbede75ecb8ee55","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-10 10:38:22.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98177d19129898e224ab6066afd8559cb7fda273","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-19 14:55:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0c032375d5ff4eab2439c405bf33324641b478e5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-19 15:37:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"56379d796d8094e178e60f3ae1102482b7f994d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 15:38:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"96ba945a1778dc8e5a1b73d8b449e3fbea111489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:48:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350a4401b9f44a70a1d6280b5eb99286\n\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/862fa08057cf4737a3465e1d9a3a65cd : SUCCESS in 4m 49s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ea4523ee0d95427698096aee397cc657 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6db991912d846cf8f5e748330a5684d : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5191be84d3864488864d14a1a36975ce : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8959712bd9d84225a5326feb6ebf31a1 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f603dfd573a74b218a59ad8d956fc239 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05832d4c30ae42bf8f18020942bd593a : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b7db5e228eb443fb04d1e5a8d2bff10 : SUCCESS in 9m 04s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/1ff51c45e11a492e8b829d55bf9ecd1f : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":3},{"id":"dc18d66e3646a69c61f985bf4d0a815098ed87f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:48:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bbe9e238d3a85d922c7fc106d72bf609f4cb2327","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:49:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66129302cc0e43b2a5b4bae042de506a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04557f47c5e42c1bf2794acc8e94772 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"92a64a342733e9be89b9102cb3f420007c0266df","revisions":{"0a31ac318eda9dafd9e96ddb0fc5261f7de5d4bf":{"kind":"REWORK","_number":1,"created":"2026-07-21 19:11:31.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/12/998212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/12/998212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/1"}}},"commit":{"parents":[{"commit":"9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8","subject":"Make block-storage endpoint optional for select commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-21 13:22:19.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-21 19:10:34.000000000","tz":-240},"subject":"Add leak-routes related options for networks","message":"Add leak-routes related options for networks\n\nThe patch adds --leak-routes for network create operation\nand --leak-routes/--no-leak-routes for the set operation\nto configure OVN BGP route leaking on networks.\n\nAssisted-By: Claude Opus 4.6 High\nRelated-Bug: #2161353\n\nChange-Id: I0cead57c3db621d25e257e4bc572aff544f9cbb7\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/0a31ac318eda9dafd9e96ddb0fc5261f7de5d4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/0a31ac318eda9dafd9e96ddb0fc5261f7de5d4bf"}]},"branch":"refs/heads/master"},"e4286537d34358288a5d98da60470569688b82b2":{"kind":"REWORK","_number":2,"created":"2026-07-23 19:51:11.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/12/998212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/12/998212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/2"}}},"commit":{"parents":[{"commit":"9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8","subject":"Make block-storage endpoint optional for select commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-21 13:22:19.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-23 19:49:32.000000000","tz":-240},"subject":"Add leak-routes related options for subnets","message":"Add leak-routes related options for subnets\n\nThe patch adds --leak-routes/--no-leak-routes for\nthe \u0027subnet set\u0027 operation to configure OVN BGP route\nleaking.\n\nAssisted-By: Claude Opus 4.6 High\nRelated-Bug: #2161353\n\nChange-Id: I0cead57c3db621d25e257e4bc572aff544f9cbb7\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e4286537d34358288a5d98da60470569688b82b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/e4286537d34358288a5d98da60470569688b82b2"}]},"branch":"refs/heads/master"},"92a64a342733e9be89b9102cb3f420007c0266df":{"kind":"REWORK","_number":3,"created":"2026-08-06 19:42:51.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/12/998212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-openstackclient","ref":"refs/changes/12/998212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-openstackclient refs/changes/12/998212/3"}}},"commit":{"parents":[{"commit":"9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8","subject":"Make block-storage endpoint optional for select commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/9726a4d7994eaee55c6e1673cb02ee7d9dc4f8d8"}]}],"author":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-21 13:22:19.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-08-06 19:41:03.000000000","tz":-240},"subject":"Add leak-routes related options for subnets","message":"Add leak-routes related options for subnets\n\nThe patch adds --leak-routes/--no-leak-routes for\nthe \u0027subnet set\u0027 operation to configure OVN BGP route\nleaking.\n\nAssisted-By: Claude Opus 4.6 High\nRelated-Bug: #2161353\n\nChange-Id: I0cead57c3db621d25e257e4bc572aff544f9cbb7\nSigned-off-by: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/92a64a342733e9be89b9102cb3f420007c0266df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-openstackclient/commit/92a64a342733e9be89b9102cb3f420007c0266df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
