)]}'
{"id":"openstack%2Fpython-searchlightclient~691032","triplet_id":"openstack%2Fpython-searchlightclient~master~Idc10f8b93260748f5886e86778f7c0f284a7ffc7","project":"openstack/python-searchlightclient","branch":"master","hashtags":[],"change_id":"Idc10f8b93260748f5886e86778f7c0f284a7ffc7","subject":"Add PDF generation support for Python-SearchlightClient","status":"ABANDONED","created":"2019-10-24 17:51:20.000000000","updated":"2019-10-25 06:59:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b49635305383358db15ed245fdd17a9b68ff34d4","_number":691032,"virtual_id_number":691032,"owner":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-24 19:56:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 19:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2956797a08d920d44ad01e00a0bb5ea780ee3b37","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 17:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f9282f71ac4d930fae7457e2a7cbb1811f5f519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:07:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5608f697ddff414799601d7c5f21527f : FAILURE in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3041c1a3fb934441bd75f9333ad789ea : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77bbefb273074c0fbd6966b1813990d1 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f66be862cb44f8fa7e6a5c06df52072 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3597f273a544714a4ded3e7c537afb6 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1f62a743cc4135b29e29b4d4ef381f : FAILURE in 3m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/162f31ffb0f7472a8085344592310724 : FAILURE in 2m 25s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/09f3b3461d754bba8f585c18e1675298 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/261c23617de74f50944e83fb9fc084bb : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ae39763d8b337adebd97aad352d13a85ec5e7442","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 18:36:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b409050de9c925b35c9f4bad59d95be7a6355a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:46:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75eb0702726446d4a8835609ccefbf6c : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b008728a39584ef98852e7d4a7d41d0f : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2202d6ca8e9c455486e377a373062ba5 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/780d7192a7a7401ba0aa5a028f966297 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2219842994b04f368649b8da4c3ef290 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9e4b013141421089d86c591ee0c7c3 : FAILURE in 3m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2a9d53d79fa466588e005a7b9de8bce : FAILURE in 3m 48s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e86e8aee47b34a42a3b7d89473a7eea7 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7789dad5f9ed448389ec172414869886 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b70ec753c07fc4e9fde3c1de3128aa91ecc61aac","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 18:59:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"775b97dbb0a38f8584674254238dacc2b8c0cf21","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 19:03:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d97a96959350b3795f2d6ec50ae39e5ac08d73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:16:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6faa630dccd04105a68358260fb4d103 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70342b12b31041c7a2a59d79ea56958d : SUCCESS in 2m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78205d7c14a4402092b33f5c6a10a11b : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebeb7c234eee40a095c41693cc3186c4 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eacf1f05e80430bb26d01344b15a46b : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24675cfba5849209efe5c9f2de3e8ef : FAILURE in 4m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/633c32296e35461888fb764c8fad670e : SUCCESS in 2m 01s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b5121384b26740f9bcff4d6483d63951 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed9bbcfdd89f4d2fb0322e31bbf28092 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d2ce726ae1f0be8f67e74d333cf40563e09b58eb","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 19:18:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9e4e250e57a59c9a4a305fb7fc231c4d4410f244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:30:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b463f17bc94b388b756656a037650c : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43f2c18303ee4429a1e05ba136dee2d0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14d11099f5ab46c4a8653d03d914f1be : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0f3fa6daf0d4627abc666aa2962d0c2 : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86e8f40ca1ae45979f897f05c0a8d653 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f47b06630e4fe584a1956c50c40a28 : FAILURE in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e4bde118d1414fffa253ed64bc3e8b3b : SUCCESS in 3m 54s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4172bed8886b473e8ab735979c18bdd4 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e8ceb7a7b7c4545a3dd1f5316171c4d : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":5},{"id":"aee633f9a7cb73530e6028dd809f2bc922dd7fee","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 19:32:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"90054a2b668b1fd32d6dc5da652fb691205bba5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:44:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68425640b7344f09a85d03f8ec3c1d6c : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/570fa1e8d8f3448ca4d0a4ec684ff791 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b89d35a849364c39b91bd56a3df1869e : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90e9a5e2da7a434ab01ed9036d015aba : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3686f918617b4f45bf5924e9c8257a11 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cd3c629e314115a281322131f16561 : FAILURE in 4m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/052486a9eeec421aa5227d7a33b732d2 : SUCCESS in 2m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1b558a5f15d94ac2ae3b8a2c638e497d : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afbc2d74999f4379b2e67363098a71bc : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":6},{"id":"71b929d0709aa252bcc9786e9040081496fbbada","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 19:49:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b21aa84c583a306e0e21042e1883020237f44d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:56:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6109e7bc8341919c5f789510ec55ea : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1d2b1cada544383a295734068f50ff1 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfdd4b64cc684188accf33ca411ecf16 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e9a650f290c44a987e9afded08bfbf6 : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/780cb45ee5ec4009a05c859a4701ca26 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9dccf346c4a4c34836a0a69c7795ded : FAILURE in 4m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5aa7e5dd62d04d56ba96aa882f00cb53 : SUCCESS in 3m 31s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/528f20956393441f9f45ee0172895398 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf1d47c4df834384a2a507bf7872334f : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":7},{"id":"b49635305383358db15ed245fdd17a9b68ff34d4","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-25 06:59:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a0a251ab9a9fdec4e9603b87b772d47c6ba75b42","revisions":{"fde884a9f15216b56b6795ddb1a086cb0a7c93f3":{"kind":"REWORK","_number":1,"created":"2019-10-24 17:51:20.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/1"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:51:18.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/fde884a9f15216b56b6795ddb1a086cb0a7c93f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/fde884a9f15216b56b6795ddb1a086cb0a7c93f3"}]},"branch":"refs/heads/master"},"668aeaeebf65b1648019e252726fd3168e205a1e":{"kind":"REWORK","_number":2,"created":"2019-10-24 18:36:16.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/2"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 18:35:54.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/668aeaeebf65b1648019e252726fd3168e205a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/668aeaeebf65b1648019e252726fd3168e205a1e"}]},"branch":"refs/heads/master"},"b97a46ecc7a28abf582debce45bed8de0f7eae0f":{"kind":"REWORK","_number":3,"created":"2019-10-24 18:59:43.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/3"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 18:59:31.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/b97a46ecc7a28abf582debce45bed8de0f7eae0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/b97a46ecc7a28abf582debce45bed8de0f7eae0f"}]},"branch":"refs/heads/master"},"63d8ed5221c7b31453497677a595b9d1f3ca265c":{"kind":"REWORK","_number":4,"created":"2019-10-24 19:03:29.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/4"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 19:03:17.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/63d8ed5221c7b31453497677a595b9d1f3ca265c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/63d8ed5221c7b31453497677a595b9d1f3ca265c"}]},"branch":"refs/heads/master"},"e6d08ea6ce5aeb172df619dabf07f3c53e1615a8":{"kind":"REWORK","_number":5,"created":"2019-10-24 19:18:49.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/5"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 19:18:37.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/e6d08ea6ce5aeb172df619dabf07f3c53e1615a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/e6d08ea6ce5aeb172df619dabf07f3c53e1615a8"}]},"branch":"refs/heads/master"},"18b866cbf46227e4e2279b118462ca4cfbb695ea":{"kind":"REWORK","_number":6,"created":"2019-10-24 19:32:42.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/6"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 19:32:28.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/18b866cbf46227e4e2279b118462ca4cfbb695ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/18b866cbf46227e4e2279b118462ca4cfbb695ea"}]},"branch":"refs/heads/master"},"a0a251ab9a9fdec4e9603b87b772d47c6ba75b42":{"kind":"REWORK","_number":7,"created":"2019-10-24 19:49:56.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/32/691032/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-searchlightclient","ref":"refs/changes/32/691032/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-searchlightclient refs/changes/32/691032/7"}}},"commit":{"parents":[{"commit":"c29a7e20a174a0491c35bb818724ba91335bdc44","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/c29a7e20a174a0491c35bb818724ba91335bdc44"}]}],"author":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 17:50:44.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 19:49:43.000000000","tz":420},"subject":"Add PDF generation support for Python-SearchlightClient","message":"Add PDF generation support for Python-SearchlightClient\n\nChange-Id: Idc10f8b93260748f5886e86778f7c0f284a7ffc7\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/a0a251ab9a9fdec4e9603b87b772d47c6ba75b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-searchlightclient/commit/a0a251ab9a9fdec4e9603b87b772d47c6ba75b42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
