)]}'
{"id":"openstack%2Fpython-senlinclient~666398","triplet_id":"openstack%2Fpython-senlinclient~master~Iadad5e6b719f3db8cad401a031718de41bfa8417","project":"openstack/python-senlinclient","branch":"master","hashtags":[],"change_id":"Iadad5e6b719f3db8cad401a031718de41bfa8417","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-19 16:56:12.000000000","updated":"2019-12-20 16:51:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f3822d0ae5c2b13b19b215c83c221cb23adbe594","_number":666398,"virtual_id_number":666398,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":-1,"date":"2019-06-25 05:43:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":-1,"date":"2019-07-14 15:11:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 05:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-14 15:11:47.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"}],"messages":[{"id":"194a70b22eb1a81a7afc80e7133ad7a5787c7ea7","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-19 16:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b377cf1a52f470d6920859adffcc70b4d6b3dc","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-19 16:57:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1891b2b4f28d4a16a0621e0a78abe16a37352d6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 17:49:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/98/666398/2/check/requirements-check/1766b84/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/666398/2/check/openstack-tox-lower-constraints/008d470/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/98/666398/2/check/openstack-tox-pep8/fe79886/ : SUCCESS in 6m 25s\n- openstack-tox-py27 http://logs.openstack.org/98/666398/2/check/openstack-tox-py27/cd07e9b/ : FAILURE in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/98/666398/2/check/openstack-tox-py36/0feadff/ : FAILURE in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/98/666398/2/check/openstack-tox-py37/bae2088/ : FAILURE in 4m 56s\n- openstackclient-check-plugins http://logs.openstack.org/98/666398/2/check/openstackclient-check-plugins/dc38093/ : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666398/2/check/openstack-tox-docs/8478c0e/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/98/666398/2/check/build-openstack-releasenotes/9547bbb/html/ : SUCCESS in 4m 35s\n- senlinclient-dsvm-functional http://logs.openstack.org/98/666398/2/check/senlinclient-dsvm-functional/8cf0d86/ : SUCCESS in 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6c6d35c03eeb88d3adb8094ed3d3d818fe22d0f","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-24 14:25:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c58f027846d91e901cc63365741230f7091fd545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:10:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/98/666398/2/check/requirements-check/14e03b5/ : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/666398/2/check/openstack-tox-lower-constraints/b93acd5/ : SUCCESS in 8m 00s\n- openstack-tox-pep8 http://logs.openstack.org/98/666398/2/check/openstack-tox-pep8/5662f4f/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/98/666398/2/check/openstack-tox-py27/59d059d/ : FAILURE in 8m 02s\n- openstack-tox-py36 http://logs.openstack.org/98/666398/2/check/openstack-tox-py36/cdcb6d5/ : FAILURE in 7m 18s\n- openstack-tox-py37 http://logs.openstack.org/98/666398/2/check/openstack-tox-py37/a78343a/ : FAILURE in 4m 55s\n- openstackclient-check-plugins http://logs.openstack.org/98/666398/2/check/openstackclient-check-plugins/6e5e4a4/ : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666398/2/check/openstack-tox-docs/e1ebd27/html/ : SUCCESS in 4m 27s\n- build-openstack-releasenotes http://logs.openstack.org/98/666398/2/check/build-openstack-releasenotes/d9efe17/html/ : SUCCESS in 4m 40s\n- senlinclient-dsvm-functional http://logs.openstack.org/98/666398/2/check/senlinclient-dsvm-functional/dde2baf/ : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4db129616a711a0d77f0035becd6d2b39832a1e0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-25 04:15:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"229494ffd03b758d737ace0867e547c6378c8b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 05:43:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/98/666398/3/check/requirements-check/0a6b152/ : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/666398/3/check/openstack-tox-lower-constraints/798dd1b/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/98/666398/3/check/openstack-tox-pep8/21b1291/ : FAILURE in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/98/666398/3/check/openstack-tox-py27/d43c074/ : FAILURE in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/98/666398/3/check/openstack-tox-py36/f42acc9/ : FAILURE in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/98/666398/3/check/openstack-tox-py37/9081e26/ : FAILURE in 3m 26s\n- openstackclient-check-plugins http://logs.openstack.org/98/666398/3/check/openstackclient-check-plugins/f80a993/ : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/98/666398/3/check/openstack-tox-docs/7809eba/html/ : SUCCESS in 2m 23s\n- build-openstack-releasenotes http://logs.openstack.org/98/666398/3/check/build-openstack-releasenotes/8259e6e/html/ : SUCCESS in 3m 00s\n- senlinclient-dsvm-functional http://logs.openstack.org/98/666398/3/check/senlinclient-dsvm-functional/71464e4/ : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d0bb5e84e9626b8c7dd4f2e5fa9454d6f1522cc","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:11:47.000000000","message":"Patch Set 3: Code-Review-1\n\nplease fix the zuul first","accounts_in_message":[],"_revision_number":3},{"id":"f3822d0ae5c2b13b19b215c83c221cb23adbe594","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-12-20 16:51:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7df6c30054395ae5b7fd25572a645f5ebd33196d","revisions":{"2c87cf6369315e6598806916b94ca5b51568f2e3":{"kind":"REWORK","_number":1,"created":"2019-06-19 16:56:12.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/98/666398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/98/666398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/1"}}},"commit":{"parents":[{"commit":"e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 16:55:51.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 16:55:51.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Update the UPPER_CONSTRAINTS_FILE for tox[1]\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iadad5e6b719f3db8cad401a031718de41bfa8417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/2c87cf6369315e6598806916b94ca5b51568f2e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/2c87cf6369315e6598806916b94ca5b51568f2e3"}]},"branch":"refs/heads/master"},"e44ff640bab3a66d51b92dfbbb7c2ba43962501e":{"kind":"REWORK","_number":2,"created":"2019-06-19 16:57:52.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/98/666398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/98/666398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/2"}}},"commit":{"parents":[{"commit":"e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 16:55:51.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 16:57:32.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Update the UPPER_CONSTRAINTS_FILE for tox[1]\n3. Remove unncessary \"\u003d\u003d\"\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iadad5e6b719f3db8cad401a031718de41bfa8417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e44ff640bab3a66d51b92dfbbb7c2ba43962501e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e44ff640bab3a66d51b92dfbbb7c2ba43962501e"}]},"branch":"refs/heads/master"},"7df6c30054395ae5b7fd25572a645f5ebd33196d":{"kind":"REWORK","_number":3,"created":"2019-06-25 04:15:39.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/98/666398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/98/666398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/98/666398/3"}}},"commit":{"parents":[{"commit":"e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e86a8cbb4bc1fa8b24a41fcab4cd5a0464c06cf7"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 16:55:51.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-25 04:15:30.000000000","tz":0},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Update the UPPER_CONSTRAINTS_FILE for tox[1]\n3. Remove unncessary \"\u003d\u003d\"\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iadad5e6b719f3db8cad401a031718de41bfa8417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/7df6c30054395ae5b7fd25572a645f5ebd33196d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/7df6c30054395ae5b7fd25572a645f5ebd33196d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
