)]}'
{"id":"openstack%2Fpython-senlinclient~689973","triplet_id":"openstack%2Fpython-senlinclient~master~I22813f6f06f80ab479338b21b4ada7ee65a12ad3","project":"openstack/python-senlinclient","branch":"master","hashtags":[],"change_id":"I22813f6f06f80ab479338b21b4ada7ee65a12ad3","subject":"Switch to Ussuri jobs","status":"ABANDONED","created":"2019-10-22 06:30:52.000000000","updated":"2021-02-15 00:00:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1823e0393ac39a0bc4c81e351558360a3a1871f8","_number":689973,"virtual_id_number":689973,"owner":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-22 07:09:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2021-02-11 18:32:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 07:09:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 18:32:35.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"10fff910f5480b0a614bd35e20e204271d14ea75","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-22 06:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0bde8f6c82fc471c454eb7aa22a771ade0ea175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 07:09:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1a31ece98ab4b46bd5159c0694a4c5e : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db291e0f0b774ffa83591279849c33f7 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eee293b6ac641b38cabacec6adfcb9e : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4622f56a836b4e44ae47129ccb910025 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c88ca978af1450d87afbb14fa5300f7 : SUCCESS in 7m 03s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e6e0505e06204cfea3fae3a53da83daf : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd326bac04340cc9d579fbd412e7229 : SUCCESS in 4m 23s\n- senlinclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/d97e55b797374898b0c4d82c4d083877 : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"597d37182505a567e671bce2058dcc050c197584","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-02-11 18:32:35.000000000","message":"Patch Set 1: Code-Review-1\n\nThis patch is not needed anymore as we are in Wallaby development cycle.\nPlease abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"1823e0393ac39a0bc4c81e351558360a3a1871f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2021-02-15 00:00:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e88ed9e70f3dfd2dc2b9436ccf50f581d146fdc0","revisions":{"e88ed9e70f3dfd2dc2b9436ccf50f581d146fdc0":{"kind":"REWORK","_number":1,"created":"2019-10-22 06:30:52.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/73/689973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/73/689973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/73/689973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/73/689973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/73/689973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/73/689973/1"}}},"commit":{"parents":[{"commit":"b1c59056bd4fe9f28afae3a7099f309dfb8a80f7","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/b1c59056bd4fe9f28afae3a7099f309dfb8a80f7"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-22 06:30:34.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-22 06:30:46.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I22813f6f06f80ab479338b21b4ada7ee65a12ad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e88ed9e70f3dfd2dc2b9436ccf50f581d146fdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/e88ed9e70f3dfd2dc2b9436ccf50f581d146fdc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
