)]}'
{"id":"openstack%2Fpython-senlinclient~691015","triplet_id":"openstack%2Fpython-senlinclient~master~I1b36e6e9b8f743308d4a8361ad1f6564a462f4f3","project":"openstack/python-senlinclient","branch":"master","topic":"node_tainted","hashtags":[],"change_id":"I1b36e6e9b8f743308d4a8361ad1f6564a462f4f3","subject":"Add support for node tainted field","status":"MERGED","created":"2019-10-24 16:35:37.000000000","updated":"2019-10-29 18:50:10.000000000","submitted":"2019-10-29 18:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691015-1572374927103-91d4b73c","meta_rev_id":"72dd75f62ec0829642d978c102b3f82040fb3694","_number":691015,"virtual_id_number":691015,"owner":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 18:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"all":[{"value":1,"date":"2019-10-24 17:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-24 17:28:00.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-25 18:02:39.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 17:26:25.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-10-25 18:02:39.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-29 18:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cf031d958f30bb939121b3af63ffc1a3fc1e019","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-24 16:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dbf748adc44bd40c8de57e17cc215e6c2268132","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-10-24 17:26:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e558765a04104ff70275c5f27615ef4583d86c9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-24 17:28:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b32f68674d1ef6a114c322cb51751899e7c37aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:28:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/072ee8e780014b859bba6c1d06a6ccf0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e78d3bacf64d7387a237ee67f29d95 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4098b2d28df4280926aad25fcd2f18c : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba0f0ba2394a4fa7a6a5f0cea27d9752 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aaba704a0be471b9f7662c96e9632d0 : SUCCESS in 6m 28s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/bdd93a0573894852b72975627bb3a1de : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/388fa80d9ac148c4ae9f011562628b83 : SUCCESS in 2m 41s\n- senlinclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f3d510c48e545cba5c8e0b37dee3133 : FAILURE in 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47cb0cda772ac6abc8b1c4e33dd6c16ae6592f94","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-25 18:02:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66103a5acfaf8b06fbf54906054f20f778cd22e5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-25 18:03:04.000000000","message":"Patch Set 1:\n\nWill be merged after the openstacksdk patch is merged.","accounts_in_message":[],"_revision_number":1},{"id":"698431c63462fb33f3aa809e68813a4b430e7ea0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-29 18:38:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10cd3006194947125e9d61116fdd77e2f1e89372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:39:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f873c89726f4740483e1b1c3890a835703f21e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb8885edae2a40b385a547598357aa49 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a94f6fe94e42349745af9cba45d376 : SUCCESS in 2m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f78d2035a47429ea284a4ae70929862 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e25ccad8aee4444ab2451cc94248e224 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce1878d73482420a9995fd74917f63d9 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8750852d59694b659b2a612efa9395c4 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"037e663480416781dba2dc137f48b8825f3406cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:48:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0276926679ba84051342b8e9b4cf0063454d07a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 18:50:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b93188fdd75455e82371b42b738d522 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c3991479862af63f4128f5cc3ff2377e2bd0b94","revisions":{"4c3991479862af63f4128f5cc3ff2377e2bd0b94":{"kind":"REWORK","_number":1,"created":"2019-10-24 16:35:37.000000000","uploader":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"ref":"refs/changes/15/691015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/15/691015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/15/691015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/15/691015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/15/691015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/15/691015/1"}}},"commit":{"parents":[{"commit":"f6453b4f0f7b5f1ff9359ca1883d3f319d028554","subject":"Return meta-data field for events list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/f6453b4f0f7b5f1ff9359ca1883d3f319d028554"}]}],"author":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-23 22:22:20.000000000","tz":0},"committer":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-24 16:35:37.000000000","tz":0},"subject":"Add support for node tainted field","message":"Add support for node tainted field\n\n- Update supported microversion to 1.13\n- Return node tainted field\n- Support setting tainted field for node update\n\nDepends-On: https://review.opendev.org/#/c/690755/\nChange-Id: I1b36e6e9b8f743308d4a8361ad1f6564a462f4f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/4c3991479862af63f4128f5cc3ff2377e2bd0b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/4c3991479862af63f4128f5cc3ff2377e2bd0b94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
