)]}'
{"id":"openstack%2Fpython-senlinclient~844652","triplet_id":"openstack%2Fpython-senlinclient~master~I4cf6becfab29f1902dfc76cb186ea53a002ac8f8","project":"openstack/python-senlinclient","branch":"master","topic":"fix-python-version","attention_set":{},"removed_from_attention_set":{"29870":{"account":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"last_update":"2022-09-20 05:56:08.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-09-20 05:45:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I4cf6becfab29f1902dfc76cb186ea53a002ac8f8","subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","status":"MERGED","created":"2022-06-04 02:33:50.000000000","updated":"2022-09-20 05:57:08.000000000","submitted":"2022-09-20 05:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844652-fix-python-version","meta_rev_id":"fb2163bfaa07f4a3b7a5f49527f283e93d237cb6","_number":844652,"virtual_id_number":844652,"owner":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 05:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-20 05:45:33.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-20 05:45:33.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-04 02:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-04 04:24:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-09-20 05:45:33.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"f0a68407944800c44e4411bcfb2dcb8c584520ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2022-06-04 02:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06a8162041fba20af6a352a578327ce4be798b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 02:50:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b723bba36277451bab2f744abc18e6cb : SUCCESS in 1m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fd934f123bb49ab90bd7f464e5cdbbe : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a46d915bf754770adaa377e47cf33d5 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b948563741c4ec4994fc46016902b65 : FAILURE in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533ee30be27841e984b07e9cf4eb83eb : SUCCESS in 4m 56s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/94fba703d83e48febf67ff4b566bcca8 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fceb22adb622489492d1661d70f32c50 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6d351c755b4b060ba53e46730b47a34777de0766","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-04 04:24:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4c9e8bb989117921f3b4fcd6c190b360fb644ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:31:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb95903ced214c4c85246ffbebad849c : SUCCESS in 2m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86cc24c141994753a7e781417a3945b0 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521ac342e4a14fe589b30de990b0fbe9 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2bfca2a3d947358825a5cd2c664dc6 : FAILURE in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2263bdba64df40e08d08bdc4b20513c5 : SUCCESS in 3m 42s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b65c680e45de49678e67c64c7bdcb8d5 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c15388b963416a8374d1b0c2c9888a : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3f0b26fe8cc2af87a187a97aa3433259acc3e659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-07 20:58:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a5c470f303d6e349019760c01c199eb3b3f20064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-07 21:16:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094e6cd9b18c45e6a63311461916a693\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/14d1490aa91145b39c5cb172b9d979e3 : SUCCESS in 2m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b4481541c7943e1b7b6f279cdc174a1 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d179f8fe8fe547c591cfa879c41cac00 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6175ef261d4ca1b022b31d6e718f24 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4e45cce85d42dfab789353935438e3 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/334d50f7102a43ecacf3c3efc95f16c2 : SUCCESS in 4m 09s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/23ed7413ad6844a89c47f2d1366a32b5 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a02f1637784764a1b6777485555b80 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a535a5846188b94e70cbc3f52370ca4b53598b98","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2022-09-12 01:58:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0b6469ab23f0b0c81758383ae14f3b63cb14bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-20 04:17:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2bb56801a7fdacc752a993e9122ab2e60f258588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 04:26:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a465151af45b4b2db8c7802c4d79912e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acb8291b7a3646589c2010bd0f994072 : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaa63f00997f477f9105cbcc2859f52c : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e0e357d7db451ebfd2915a8c634f3a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ba66f055078458db8cf1c7db5b98345 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56b4f809577d49f995dbcad1c2f7bf51 : SUCCESS in 4m 29s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7d10c51ec196446eb0d34d8a2162f61e : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c3287f12f84b70b026129c0442af43 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"17c6069d002b97ac6083034b9b975c968d0d69d0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-20 05:45:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"02eeac6e3e6b6baa0b3bb83044c37bc2a657374b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 05:45:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f27ba8b7538c6115243813f00d0644c0a41683a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 05:56:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0eee9700a94437b659337d24dc31f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5739736d49c744b3a3b0d3bd858f31e9 : SUCCESS in 2m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a1d9a33c8fb40298cce5af3af5bd8f1 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77160cdafc194ae2b9be05a5007d1153 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443b18ca231549e78a189e968ec4b04b : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e032422e3d045f49811bb0a1ef37de9 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe1dbd42303472ca3a3b036386f83a2 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"aa26c73e3ab5f84fe64bdd207d1994de548049dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 05:56:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fb2163bfaa07f4a3b7a5f49527f283e93d237cb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 05:57:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d63005bde9d64af79778e932a80fc72e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950c181ad5eb4b8faddbc03658e9387a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"681b5de6b58bf84337d9b20094dc4d7e445e162e","revisions":{"690a5336315c1877633afa682072804abbbfe74e":{"kind":"REWORK","_number":1,"created":"2022-06-04 02:33:50.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/52/844652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/52/844652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/1"}}},"commit":{"parents":[{"commit":"36a61a7912e4a28558058425eeefe37805d1f4b5","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/36a61a7912e4a28558058425eeefe37805d1f4b5"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2022-06-04 02:33:04.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2022-06-04 02:33:04.000000000","tz":480},"subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","message":"The Python 3.6 and Python 3.7 Support has been dropped since zed\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I4cf6becfab29f1902dfc76cb186ea53a002ac8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/690a5336315c1877633afa682072804abbbfe74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/690a5336315c1877633afa682072804abbbfe74e"}]},"branch":"refs/heads/master"},"4e80df756e02b222660f9d29da988bbac79ad722":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-07 20:58:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/52/844652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/52/844652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/2"}}},"commit":{"parents":[{"commit":"c17a8c48f29d88dcd4e50254e31a7e70337a2139","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/c17a8c48f29d88dcd4e50254e31a7e70337a2139"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2022-06-04 02:33:04.000000000","tz":480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-07 20:58:10.000000000","tz":0},"subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","message":"The Python 3.6 and Python 3.7 Support has been dropped since zed\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I4cf6becfab29f1902dfc76cb186ea53a002ac8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/4e80df756e02b222660f9d29da988bbac79ad722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/4e80df756e02b222660f9d29da988bbac79ad722"}]},"branch":"refs/heads/master","description":"Rebase"},"681b5de6b58bf84337d9b20094dc4d7e445e162e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-20 04:17:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/52/844652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/52/844652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/52/844652/3"}}},"commit":{"parents":[{"commit":"a91bb185faef12da36d72802b7294bf784ce4d4b","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/a91bb185faef12da36d72802b7294bf784ce4d4b"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2022-06-04 02:33:04.000000000","tz":480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-09-20 04:17:10.000000000","tz":0},"subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","message":"The Python 3.6 and Python 3.7 Support has been dropped since zed\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I4cf6becfab29f1902dfc76cb186ea53a002ac8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/681b5de6b58bf84337d9b20094dc4d7e445e162e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/681b5de6b58bf84337d9b20094dc4d7e445e162e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[]}
