)]}'
{"id":"openstack%2Fpython-senlinclient~868968","triplet_id":"openstack%2Fpython-senlinclient~master~I778226115c587875e0db995b2641af74a2854015","project":"openstack/python-senlinclient","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"29870":{"account":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"last_update":"2023-02-16 17:58:43.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-02-16 17:47:23.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I778226115c587875e0db995b2641af74a2854015","subject":"Fix tox.ini with tox4","status":"MERGED","created":"2023-01-01 06:45:19.000000000","updated":"2023-02-16 18:00:50.000000000","submitted":"2023-02-16 17:58:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"868968-tox4","meta_rev_id":"a7071ee24f01df3643cf2543670ff6fc350af6e7","_number":868968,"virtual_id_number":868968,"owner":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 17:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2023-02-16 17:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-16 05:25:36.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-16 17:47:23.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-01 07:28:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 19:06:27.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-02-16 05:25:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"1d50a31290efa2748e797423513fbb328dca589e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2023-01-01 06:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"771f404b96831dabc42663e4fd092d5428b3ef02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 07:28:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/016508e011cc4469a6940fbc7e1ec690\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb6416c2f44347ae96aa803add9494af : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76deb1362d7c452c9fc6f68cd79a8d5a : FAILURE in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4905c33bdbe427aa5a3efe2ec6ff1e1 : FAILURE in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d11b3886ba9487582eca5ce3404abe5 : FAILURE in 3m 07s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7990362eccde4fa4a1ade4f6ef9e63b7 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65105b2a7d8f47558a16af325930f680 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d95b202af1a46d388735ecb720e7b0d : SUCCESS in 3m 30s\n- senlinclient-functional https://zuul.opendev.org/t/openstack/build/ed7de9f912374b32a7c56137cfaad32f : SUCCESS in 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cf403b54c2eda0e17bab72d5ce0908ffe0773f8","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2023-01-05 08:40:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d2c0833363da5f3f14fa1701f4ff139ab22359c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 09:11:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c55043dfad88431197e0679ec1b2f4ff\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e64a695870044631a939122e9455ba18 : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc420d157c924316b939372b349bd6c4 : FAILURE in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27cab8e55b10457b904d0967daf29633 : FAILURE in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1180914d3bfb47e788f5fa5d6c59c8a0 : FAILURE in 2m 50s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/75559be10999485a98ef7d14f202ec9f : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7ac5eb469e48bdbc29eb05012d73b2 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72bda24f70534315a7cfb349fee51b24 : SUCCESS in 2m 50s\n- senlinclient-functional https://zuul.opendev.org/t/openstack/build/46c0e879a45b4e23b055932dae126ddd : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7391e31d4355596a2dc533f2ed657e0690c9c23","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-13 19:06:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"59c7c290cfacf0221f42f620d27adedea3a43df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-16 05:23:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8eeeebe90e4ea4f41ab01f65dca9c7dc97522084","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-16 05:25:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"80385bdc2a104073d75e6e4716041c106377c622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:04:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd68e5f1c2645a18105fd949df9c496\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780786939bac43088a5a949004e25d68 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb7c3dc2f1c044d192c4350d4fcc6ea3 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99ebddb237b94aa19b57411dad67a4f7 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29368a52274b4211a2f41c2e78fb6249 : SUCCESS in 3m 02s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/af26ee892c34485d9244adb7c8e4294e : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc3067f32334cd3899861f4d8d586f1 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/874dfaa5102442b392789f13445c81e3 : SUCCESS in 3m 08s\n- senlinclient-functional https://zuul.opendev.org/t/openstack/build/fc38da40865b4735a432e87349ac7b60 : SUCCESS in 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d62ea950c136a1782812b57081d9d6900763c80d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-16 17:47:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"883361447e483416868f09754c00c0475a5a74c3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-16 17:47:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9196bf7c87de590a1cabbb4670e76b06a0d54c2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 17:47:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c629fde39008b8b406e8a4c821c9ce22809e8d29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-16 17:47:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db5d73095a0c36b83aa0357510a95117cdb1a105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 17:58:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9597c3c132434d9b2f5873e066d189\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eebcdecbb12547d5abf9b20b0667161b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7a258ad7364fd99d1b9ffabc3236bd : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f46b8ef559c48b09af77c2a14b1e59a : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17990770b4174f819958c74ed17d351d : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b4783edf85469ca1b033395b466dcf : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/839bf093050042b5903eef176658130e : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fa7d7ef96d31beb80c15679f67cc4038de2af669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 17:58:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7071ee24f01df3643cf2543670ff6fc350af6e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:00:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ae0e6603984785a5bcfc62132f12e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc7fe566d2b4f929ff6ef6d751a343e : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91b01db5538f4491a151924ee93259f4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98e079f0d362d35bdf06a686476d6b24dd3fe433","revisions":{"913207b8ce32e9597ec4e8429ca730ab25ac6d0a":{"kind":"REWORK","_number":1,"created":"2023-01-01 06:45:19.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/68/868968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/68/868968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/1"}}},"commit":{"parents":[{"commit":"681b5de6b58bf84337d9b20094dc4d7e445e162e","subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/681b5de6b58bf84337d9b20094dc4d7e445e162e"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2023-01-01 06:42:57.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2023-01-01 06:42:57.000000000","tz":480},"subject":"Fix tox.ini with tox4","message":"Fix tox.ini with tox4\n\nskipsdist setting breaks the docs environment\n\nChange-Id: I778226115c587875e0db995b2641af74a2854015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/913207b8ce32e9597ec4e8429ca730ab25ac6d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/913207b8ce32e9597ec4e8429ca730ab25ac6d0a"}]},"branch":"refs/heads/master"},"98e079f0d362d35bdf06a686476d6b24dd3fe433":{"kind":"REWORK","_number":2,"created":"2023-02-16 05:23:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/868968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-senlinclient","ref":"refs/changes/68/868968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-senlinclient refs/changes/68/868968/2"}}},"commit":{"parents":[{"commit":"681b5de6b58bf84337d9b20094dc4d7e445e162e","subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/681b5de6b58bf84337d9b20094dc4d7e445e162e"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2023-01-01 06:42:57.000000000","tz":480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-16 05:23:34.000000000","tz":-480},"subject":"Fix tox.ini with tox4","message":"Fix tox.ini with tox4\n\nskipsdist setting breaks the docs environment\n\nChange-Id: I778226115c587875e0db995b2641af74a2854015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/98e079f0d362d35bdf06a686476d6b24dd3fe433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-senlinclient/commit/98e079f0d362d35bdf06a686476d6b24dd3fe433"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[]}
