)]}'
{"id":"openstack%2Fpython-tackerclient~856267","triplet_id":"openstack%2Fpython-tackerclient~master~Ide4d8c8652cdf8ac611e65376a21806488667ace","project":"openstack/python-tackerclient","branch":"master","topic":"bp/support-auto-lcm/doc","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-09-12 09:03:48.000000000","reason":"\u003cGERRIT_ACCOUNT_31857\u003e replied on the change","reason_account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-09-12 06:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_25701\u003e replied on the change","reason_account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}},"35016":{"account":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"last_update":"2022-09-12 09:15:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide4d8c8652cdf8ac611e65376a21806488667ace","subject":"Add docs for cnf auto heal and scale","status":"MERGED","created":"2022-09-07 11:45:11.000000000","updated":"2022-09-12 09:16:49.000000000","submitted":"2022-09-12 09:15:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"856267-bp/support-auto-lcm/doc","meta_rev_id":"4349a6273c80f5bf1fc39fa7c4f2b8b74a5af7dd","_number":856267,"virtual_id_number":856267,"owner":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 09:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":2,"date":"2022-09-12 09:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":2,"date":"2022-09-12 06:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"all":[{"value":1,"date":"2022-09-12 09:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 11:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 14:16:04.000000000","updated_by":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"reviewer":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"state":"REVIEWER"},{"updated":"2022-09-07 14:16:04.000000000","updated_by":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"488d642b9fb53ef2584dc38899150c272a00aed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"date":"2022-09-07 11:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acacf81235fc6be9f8111e035207e177ef75366f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 11:55:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9210bf41862461aa7d15286a9769305\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aea349a0c484ba39bef3c24ad09c42f : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881bd9bf259f41ecaa378476a8074695 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6ae4da529d679cae86f4ddc378b3260dd31a9879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"date":"2022-09-07 14:14:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d1357059fd9dca48815c93137cb585ee1fe2db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"date":"2022-09-07 14:14:42.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a853ad2c91aacda4b304ffdc71dc3ec152206a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 14:20:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b566a51035c44549c2fdbb80a6c3837\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa5782a73014f789d1d2502933b7025 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecafc9fd5d9a4d1cb5c9d0e704106952 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ab4935cc03b8b4152f16b438651cb1732127236e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-09-07 19:10:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d6ba2837ba5a897677bf829aa665c245b609ea39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"date":"2022-09-08 10:49:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b4811712c2c8d40db08b105abcf2af1e4830270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:02:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2399133cd8fe433d83276cbc2610aff3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d8a7d683034d9ba93107d17d3f0626 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43e156fe18542b6a3fb6ab99a9d6535 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"a9dc51d90440060a7a55d04526425d82c25b2578","author":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"date":"2022-09-12 01:59:15.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7b927ba71ef672d11ba0e77f18319b17a0aadb0e","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-09-12 06:55:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"12493690a5614a32fe01142f2ae4576127b8dd8a","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-09-12 09:03:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bae3cf7a5ae8f774e6816bd4e15cef56b5491815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:04:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"39f48f3e22d4a47bebb3d623281967d8795459a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:15:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3a52d7392d4b29b4d2dec9e21b42d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacb87089f4945c4abc2db2b3f216797 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44024824ee304432a8f724b044597191 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"32c8fac6259ba6898669dc474843a72fb48662a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:15:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4349a6273c80f5bf1fc39fa7c4f2b8b74a5af7dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:16:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e7c2cc75a849f29a7660c05a09a577\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325334ba1c9e43b38619d49fdabea8cb : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"80ee72126392070993082bad73422fd64713c6f0","revisions":{"3c67f3443ce1f4ee805e9168b790532874a864f5":{"kind":"REWORK","_number":1,"created":"2022-09-07 11:45:11.000000000","uploader":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"ref":"refs/changes/67/856267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tackerclient","ref":"refs/changes/67/856267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/1"}}},"commit":{"parents":[{"commit":"64d7a876703b98f6fb83d413a8c5c801a5e752d4","subject":"Add support cnf auto heal and scale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/64d7a876703b98f6fb83d413a8c5c801a5e752d4"}]}],"author":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 11:44:23.000000000","tz":540},"committer":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 11:44:23.000000000","tz":540},"subject":"[WIP] Update API doc and CLI doc","message":"[WIP] Update API doc and CLI doc\n\nThis patch makes the following updates to the API and CLI docs.\n\nAPI doc:\n* Add support cnf auto heal and scale\n\nCLI doc:\n* Add support cnf auto heal and scale\n\nImplements: blueprint support-auto-lcm\nChange-Id: Ide4d8c8652cdf8ac611e65376a21806488667ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/3c67f3443ce1f4ee805e9168b790532874a864f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/3c67f3443ce1f4ee805e9168b790532874a864f5"}]},"branch":"refs/heads/master"},"b0bfafd57c419c1a23a97f19468790cc25b6e039":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-07 14:14:06.000000000","uploader":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"ref":"refs/changes/67/856267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tackerclient","ref":"refs/changes/67/856267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/2"}}},"commit":{"parents":[{"commit":"64d7a876703b98f6fb83d413a8c5c801a5e752d4","subject":"Add support cnf auto heal and scale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/64d7a876703b98f6fb83d413a8c5c801a5e752d4"}]}],"author":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 11:44:23.000000000","tz":540},"committer":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 14:14:06.000000000","tz":0},"subject":"Update API doc and CLI doc","message":"Update API doc and CLI doc\n\nThis patch makes the following updates to the API and CLI docs.\n\nAPI doc:\n* Add support cnf auto heal and scale\n\nCLI doc:\n* Add support cnf auto heal and scale\n\nImplements: blueprint support-auto-lcm\nChange-Id: Ide4d8c8652cdf8ac611e65376a21806488667ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/b0bfafd57c419c1a23a97f19468790cc25b6e039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/b0bfafd57c419c1a23a97f19468790cc25b6e039"}]},"branch":"refs/heads/master","description":"Edit commit message"},"042ad0c987241b1a21de671f9d5504d99f513912":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-07 14:14:42.000000000","uploader":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"ref":"refs/changes/67/856267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tackerclient","ref":"refs/changes/67/856267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/3"}}},"commit":{"parents":[{"commit":"64d7a876703b98f6fb83d413a8c5c801a5e752d4","subject":"Add support cnf auto heal and scale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/64d7a876703b98f6fb83d413a8c5c801a5e752d4"}]}],"author":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 11:44:23.000000000","tz":540},"committer":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 14:14:42.000000000","tz":0},"subject":"Add docs for cnf auto heal and scale","message":"Add docs for cnf auto heal and scale\n\nThis patch makes the following updates to the API and CLI docs.\n\nAPI doc:\n* Add support cnf auto heal and scale\n\nCLI doc:\n* Add support cnf auto heal and scale\n\nImplements: blueprint support-auto-lcm\nChange-Id: Ide4d8c8652cdf8ac611e65376a21806488667ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/042ad0c987241b1a21de671f9d5504d99f513912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/042ad0c987241b1a21de671f9d5504d99f513912"}]},"branch":"refs/heads/master","description":"Edit commit message"},"80ee72126392070993082bad73422fd64713c6f0":{"kind":"REWORK","_number":4,"created":"2022-09-08 10:49:58.000000000","uploader":{"_account_id":35016,"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","username":"shimizu-koji"},"ref":"refs/changes/67/856267/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tackerclient","ref":"refs/changes/67/856267/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tackerclient refs/changes/67/856267/4"}}},"commit":{"parents":[{"commit":"64d7a876703b98f6fb83d413a8c5c801a5e752d4","subject":"Add support cnf auto heal and scale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/64d7a876703b98f6fb83d413a8c5c801a5e752d4"}]}],"author":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-07 11:44:23.000000000","tz":540},"committer":{"name":"Koji Shimizu","email":"shimizu.koji@fujitsu.com","date":"2022-09-08 10:49:24.000000000","tz":540},"subject":"Add docs for cnf auto heal and scale","message":"Add docs for cnf auto heal and scale\n\nThis patch makes the following updates to the API and CLI docs.\n\n* Add support cnf auto heal and scale\n\nImplements: blueprint support-auto-lcm\nChange-Id: Ide4d8c8652cdf8ac611e65376a21806488667ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/80ee72126392070993082bad73422fd64713c6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tackerclient/commit/80ee72126392070993082bad73422fd64713c6f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"}}]}],"submit_requirements":[]}
