)]}'
{"id":"openstack%2Fpython-tripleoclient~575003","triplet_id":"openstack%2Fpython-tripleoclient~master~I2867598f717b3126071e77a7826f48f6c7584ce2","project":"openstack/python-tripleoclient","branch":"master","topic":"bug/1758906","hashtags":[],"change_id":"I2867598f717b3126071e77a7826f48f6c7584ce2","subject":"Revert \"Revert \"Alter default http boot path for containerized Ironic\"\"","status":"MERGED","created":"2018-06-13 06:21:44.000000000","updated":"2018-07-18 05:16:57.000000000","submitted":"2018-07-18 05:16:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"575003-1531891017352-8bcb5b06","meta_rev_id":"c39f036aa9cec705e031640406fb4e206cbb4508","_number":575003,"virtual_id_number":575003,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2018-07-18 05:16:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-07-17 03:47:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2018-07-17 06:57:35.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2018-07-17 01:26:25.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2018-07-17 05:33:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2018-07-17 08:30:36.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"reviewer":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"state":"REVIEWER"},{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"reviewer":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"state":"REVIEWER"},{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-06-13 06:21:44.000000000","updated_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2018-06-18 09:28:03.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2018-07-17 01:26:25.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-07-17 03:47:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-17 05:33:03.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2018-07-17 06:57:35.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2018-07-18 05:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fc9eddc2280c3258a318cb3585a86b7a9146337","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-06-13 06:21:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8054473e327c891c61c6375bb131b3ab4d3bf2b6","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-06-13 06:24:22.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s wait for https://review.openstack.org/#/c/566916/","accounts_in_message":[],"_revision_number":1},{"id":"c8477f3634b91646ba5b9567ffcaf158b1ef0dde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 09:18:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/575003/1/check/openstack-tox-pep8/1a16d4e/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/03/575003/1/check/openstack-tox-py27/deae8e2/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/03/575003/1/check/openstack-tox-py35/ee02106/ : SUCCESS in 4m 22s\n- openstackclient-check-plugins http://logs.openstack.org/03/575003/1/check/openstackclient-check-plugins/133035c/ : SUCCESS in 7m 42s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/03/575003/1/check/build-openstack-releasenotes/5a638dc/html/ : SUCCESS in 3m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/575003/1/check/build-openstack-sphinx-docs/e0a6306/html/ : SUCCESS in 4m 29s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-undercloud-oooq/b452342/ : SUCCESS in 1h 30m 12s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-undercloud-upgrades/c1d83a3/ : SUCCESS in 1h 26m 20s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-undercloud-containers/ba6b219/ : SUCCESS in 1h 24m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1005bf0/ : SUCCESS in 2h 01m 49s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-3nodes-multinode/f13f065/ : SUCCESS in 2h 15m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-containers-multinode/db4381a/ : SUCCESS in 2h 25m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7e7f535/ : SUCCESS in 2h 28m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8142026/ : SUCCESS in 2h 40m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/03/575003/1/check/tripleo-ci-centos-7-standalone/9cd6be2/ : SUCCESS in 47m 20s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/03/575003/1/check/tripleo-buildimage-overcloud-full-centos-7/991f77e/ : SUCCESS in 23m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/575003/1/check/openstack-tox-lower-constraints/94caaea/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2c28f18f83cb61ce32a91648592f34a84af48e71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-06-13 11:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/15970/ : FAILURE in 4h 41m 02s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/15458/ : SUCCESS in 2h 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6d9e8862e202555a3dfb605faa9aeb468775d9ab","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-06-18 09:28:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"904e4072d97855dff078f902c934c5b66af4b419","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-07-04 11:55:48.000000000","message":"Patch Set 1:\n\ncheck-rdo\n\nhttps://review.openstack.org/#/c/566916/ has landed","accounts_in_message":[],"_revision_number":1},{"id":"3fddf10d2f234b62579c6378936d724770c7f3c6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-04 15:35:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/03/575003/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/00c62c5/ : SUCCESS in 2h 57m 27s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/03/575003/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/913a3d2/ : SUCCESS in 2h 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ab335350ab8a75f03132ffad63a92cacc310dc95","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2018-07-11 15:01:33.000000000","message":"Patch Set 1: Code-Review+1\n\nHolding patch has been merged, guess we can revive this one, Bogdan?","accounts_in_message":[],"_revision_number":1},{"id":"86a38dd81b4fda97e8db369a51977ab6b5ef14a3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-07-16 22:32:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8094a6c5089105d169c46a2f621a9e8ab3af618","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-07-16 22:35:06.000000000","message":"Patch Set 2:\n\nI had to update unit tests, please review again.","accounts_in_message":[],"_revision_number":2},{"id":"385f2b7762c91fc8a5935b1630d5b935b94f0a58","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-07-17 01:26:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bcf346d004c7a6074c51222a29f84469c753f84c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 02:15:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/575003/2/check/openstack-tox-pep8/1243797/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/03/575003/2/check/openstack-tox-py27/5eee05e/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/03/575003/2/check/openstack-tox-py35/adb41df/ : SUCCESS in 5m 43s\n- openstackclient-check-plugins http://logs.openstack.org/03/575003/2/check/openstackclient-check-plugins/d297669/ : SUCCESS in 5m 57s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/03/575003/2/check/build-openstack-releasenotes/45405ab/html/ : SUCCESS in 3m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/575003/2/check/build-openstack-sphinx-docs/4dfd367/html/ : SUCCESS in 5m 34s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-oooq/a8e67b4/ : SUCCESS in 1h 28m 45s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-upgrades/a0ff80f/ : SUCCESS in 1h 19m 09s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-containers/a960da8/ : SUCCESS in 1h 28m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0d1d8a6/ : SUCCESS in 1h 54m 44s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-3nodes-multinode/5965f9a/ : SUCCESS in 2h 22m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-containers-multinode/db43861/ : SUCCESS in 2h 11m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/524b792/ : SUCCESS in 2h 23m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1a13e7e/ : SUCCESS in 2h 08m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-standalone/930580c/ : SUCCESS in 1h 05m 09s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/03/575003/2/check/tripleo-buildimage-overcloud-full-centos-7/9001572/ : SUCCESS in 23m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/575003/2/check/openstack-tox-lower-constraints/c9d32aa/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a9e2d192215c3a713e50dd0bbaa9316640beed2a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-17 03:47:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/03/575003/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/29fe2fb/ : SUCCESS in 2h 22m 12s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/03/575003/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/717a651/ : SUCCESS in 2h 23m 07s","accounts_in_message":[],"_revision_number":2},{"id":"232d01d85ebca9c9de54b0dbaff69a209ce6d038","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2018-07-17 05:33:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a3141aa886c628c27fbd5005659ec6d4ae52a160","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2018-07-17 06:57:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df2507fa1377190a00df97d8ccf731267317f9eb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-07-17 08:30:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb0626d3781442b0a491e599e11d926319a29387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 08:30:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45f9332de04ea4b5a1d5057f0daa2c1a92304114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 13:30:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/03/575003/2/gate/openstack-tox-pep8/a1ffcaf/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/03/575003/2/gate/openstack-tox-py27/ab26c2b/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/03/575003/2/gate/openstack-tox-py35/880db4f/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/03/575003/2/gate/build-openstack-releasenotes/c876213/html/ : SUCCESS in 3m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/575003/2/gate/build-openstack-sphinx-docs/6a2017c/html/ : SUCCESS in 4m 40s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-undercloud-oooq/7abd77f/ : SUCCESS in 1h 24m 06s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-undercloud-containers/568f41e/ : SUCCESS in 2h 30m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-containerized-undercloud-upgrades/b9ae387/ : SUCCESS in 2h 08m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-containers-multinode/3f6a0b2/ : SUCCESS in 2h 45m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/438dc8b/ : TIMED_OUT in 3h 02m 01s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/03/575003/2/gate/tripleo-buildimage-overcloud-full-centos-7/5ecb7a9/ : SUCCESS in 26m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/575003/2/gate/openstack-tox-lower-constraints/7d3e529/ : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6d2d84041fd20342d167d61f531073bab742cee4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-07-17 19:21:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"facece42781f3b9910c8518d3b469b965a573c16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-17 21:52:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/03/575003/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/df1339c/ : SUCCESS in 2h 25m 57s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/03/575003/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/aa43e41/ : SUCCESS in 2h 27m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c188f0ddcbf85df6e6e64aaf489f731d6d788afa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 22:01:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/575003/2/check/openstack-tox-pep8/a4e59ce/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/03/575003/2/check/openstack-tox-py27/1f423e2/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/03/575003/2/check/openstack-tox-py35/df42d2c/ : SUCCESS in 4m 18s\n- openstackclient-check-plugins http://logs.openstack.org/03/575003/2/check/openstackclient-check-plugins/54f4645/ : SUCCESS in 6m 09s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/03/575003/2/check/build-openstack-releasenotes/70a48a8/html/ : SUCCESS in 3m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/575003/2/check/build-openstack-sphinx-docs/d286a8e/html/ : SUCCESS in 3m 53s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-oooq/9365a2f/ : SUCCESS in 1h 09m 17s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-upgrades/a618615/ : SUCCESS in 1h 09m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-undercloud-containers/c520e50/ : SUCCESS in 1h 25m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7e47a33/ : SUCCESS in 1h 36m 47s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-3nodes-multinode/8c87e59/ : SUCCESS in 2h 30m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-containers-multinode/5470ebd/ : SUCCESS in 2h 10m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e77cd6e/ : SUCCESS in 2h 30m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3cf748d/ : SUCCESS in 1h 55m 40s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/03/575003/2/check/tripleo-ci-centos-7-standalone/5dbd6ae/ : SUCCESS in 52m 08s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/03/575003/2/check/tripleo-buildimage-overcloud-full-centos-7/b68648e/ : SUCCESS in 26m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/575003/2/check/openstack-tox-lower-constraints/ba1c158/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3104b5b6aa3dd324c0488168a24619ed09f78720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 22:02:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2bc04a8c1455c2a80711cc88b43d08cd98c213c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-18 05:16:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/575003/2/gate/openstack-tox-pep8/c559c95/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/03/575003/2/gate/openstack-tox-py27/611675a/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/03/575003/2/gate/openstack-tox-py35/17dea41/ : SUCCESS in 5m 19s\n- build-openstack-releasenotes http://logs.openstack.org/03/575003/2/gate/build-openstack-releasenotes/4a4a4e9/html/ : SUCCESS in 4m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/575003/2/gate/build-openstack-sphinx-docs/fd4f674/html/ : SUCCESS in 3m 49s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-undercloud-oooq/020b91f/ : SUCCESS in 1h 13m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-undercloud-containers/9800cd7/ : SUCCESS in 1h 35m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-containerized-undercloud-upgrades/60b9a9c/ : SUCCESS in 1h 41m 21s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-containers-multinode/ddc0eed/ : SUCCESS in 2h 12m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/03/575003/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e1237b9/ : SUCCESS in 2h 17m 00s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/03/575003/2/gate/tripleo-buildimage-overcloud-full-centos-7/f314bba/ : SUCCESS in 25m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/575003/2/gate/openstack-tox-lower-constraints/4e57b62/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"674ae1a0d61279ac50b0dd57f6e66b77fc74e5d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-18 05:16:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e2af1aeb31be7168780d2bc053210a94f5fcc55","revisions":{"a1fa2d68ca56d5a311928487f20f9b797c97c660":{"kind":"REWORK","_number":1,"created":"2018-06-13 06:21:44.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/03/575003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/03/575003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/1"}}},"commit":{"parents":[{"commit":"9ac6bd66af06bfb015a4f903a412b0eadee9e153","subject":"Revert \"Alter default http boot path for containerized Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9ac6bd66af06bfb015a4f903a412b0eadee9e153"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2018-06-13 06:21:44.000000000","tz":0},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2018-06-13 06:21:44.000000000","tz":0},"subject":"Revert \"Revert \"Alter default http boot path for containerized Ironic\"\"","message":"Revert \"Revert \"Alter default http boot path for containerized Ironic\"\"\n\nThis reverts commit 9ac6bd66af06bfb015a4f903a412b0eadee9e153.\n\nChange-Id: I2867598f717b3126071e77a7826f48f6c7584ce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a1fa2d68ca56d5a311928487f20f9b797c97c660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a1fa2d68ca56d5a311928487f20f9b797c97c660"}]},"branch":"refs/heads/master"},"1e2af1aeb31be7168780d2bc053210a94f5fcc55":{"kind":"REWORK","_number":2,"created":"2018-07-16 22:32:41.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/03/575003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/03/575003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/03/575003/2"}}},"commit":{"parents":[{"commit":"300dfb389ec7030ab4b4894e167514bb1a8c042e","subject":"Merge \"Respect THT_HOME for vstate dropin file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/300dfb389ec7030ab4b4894e167514bb1a8c042e"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2018-06-13 06:21:44.000000000","tz":0},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2018-07-16 22:32:35.000000000","tz":-240},"subject":"Revert \"Revert \"Alter default http boot path for containerized Ironic\"\"","message":"Revert \"Revert \"Alter default http boot path for containerized Ironic\"\"\n\nThis reverts commit 9ac6bd66af06bfb015a4f903a412b0eadee9e153.\n\nChange-Id: I2867598f717b3126071e77a7826f48f6c7584ce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1e2af1aeb31be7168780d2bc053210a94f5fcc55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1e2af1aeb31be7168780d2bc053210a94f5fcc55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
