)]}'
{"id":"openstack%2Fpython-tripleoclient~587305","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907","project":"openstack/python-tripleoclient","branch":"stable/queens","topic":"queens/multiarch-support","hashtags":[],"change_id":"Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907","subject":"Allow passing an agent-name as a command line argument","status":"MERGED","created":"2018-07-31 07:21:08.000000000","updated":"2019-09-07 01:30:26.000000000","submitted":"2019-09-07 01:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"587305-1567819826792-b79a56ac","meta_rev_id":"66b15dc575a4a8c36445f2e57a95f3a2b7212d75","_number":587305,"virtual_id_number":587305,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-09-07 01:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-29 23:13:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":2,"date":"2019-09-03 16:38:33.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-09-06 22:08:01.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-09-06 22:08:01.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 15:15:48.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-15 08:13:23.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-08-21 18:56:15.000000000","updated_by":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"reviewer":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"state":"REVIEWER"},{"updated":"2019-08-29 23:13:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 16:38:33.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-09-06 22:08:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-07 01:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bba7bfa48c8dce1fb3f256eaec473d536dfb904e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 07:21:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcc2f91d290b6341bf7e78915e7ceada3a3ce992","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 08:00:26.000000000","message":"Patch Set 1: Workflow-1\n\nInitial backport currently untested","accounts_in_message":[],"_revision_number":1},{"id":"0dc5f9a306edfc4589db344d330bc199dc3b317c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-31 09:31:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens http://logs.rdoproject.org/05/587305/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens/d2ebe7b/ : SUCCESS in 2h 03m 01s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens http://logs.rdoproject.org/05/587305/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens/9bc2309/ : FAILURE in 1h 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"21202ddace50b8bfb98f8f7af3657295c8678679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 10:28:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/1/check/openstack-tox-pep8/874f548/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/1/check/openstack-tox-py27/15408e0/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/1/check/openstack-tox-py35/ccabd07/ : SUCCESS in 4m 26s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/1/check/openstackclient-check-plugins/f57a137/ : FAILURE in 7m 05s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/1/check/build-openstack-sphinx-docs/6cf10df/html/ : SUCCESS in 3m 18s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-oooq/09d6141/ : SUCCESS in 1h 05m 50s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-upgrades/876f5d2/ : SUCCESS in 1h 19m 54s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-3nodes-multinode/a892228/ : FAILURE in 1h 28m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-containers-multinode/acce66b/ : SUCCESS in 2h 10m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a0f3706/ : TIMED_OUT in 3h 00m 47s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/1/check/tripleo-buildimage-overcloud-full-centos-7/93398c9/ : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2ad1e44f3f117caa2d519a9a3b616133b8719ce1","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-08-02 23:19:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2565840356693bf2ab4cab1a52f66344e2b8886","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-02 23:55:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens http://logs.rdoproject.org/05/587305/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens/40a135b/ : FAILURE in 29m 26s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens http://logs.rdoproject.org/05/587305/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens/442195a/ : FAILURE in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3d007671edc380fae139c0055e703d7a489f49f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 02:29:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/1/check/openstack-tox-pep8/f50dca6/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/1/check/openstack-tox-py27/90953ba/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/1/check/openstack-tox-py35/3212bd0/ : SUCCESS in 3m 57s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/1/check/openstackclient-check-plugins/710c6d5/ : FAILURE in 5m 44s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/1/check/build-openstack-sphinx-docs/8f4f182/html/ : SUCCESS in 3m 13s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-oooq/fbd5caf/ : SUCCESS in 1h 22m 23s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-upgrades/b7e22c8/ : SUCCESS in 1h 04m 05s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-3nodes-multinode/2e46e93/ : FAILURE in 1h 14m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-containers-multinode/882920a/ : SUCCESS in 2h 24m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f320a64/ : TIMED_OUT in 3h 08m 21s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/1/check/tripleo-buildimage-overcloud-full-centos-7/0e59036/ : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":1},{"id":"275de34309732c1f8bff1b8eaf272bf77609c76d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-10-23 23:09:00.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8a45f99ec6701e408de9cf634726be0e72f597f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 02:08:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-oooq/13d7d53/ : SUCCESS in 1h 12m 42s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-undercloud-upgrades/5cdfe7f/ : SUCCESS in 1h 22m 11s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-containers-multinode/ea123a1/ : SUCCESS in 2h 22m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/812cd17/ : FAILURE in 2h 50m 37s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/1/check/tripleo-buildimage-overcloud-full-centos-7/d1dbef8/ : SUCCESS in 29m 30s\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/1/check/openstack-tox-pep8/9281be6/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/1/check/openstack-tox-py27/13b3b9e/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/1/check/openstack-tox-py35/c3379ea/ : SUCCESS in 4m 59s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/1/check/openstackclient-check-plugins/6095fd1/ : FAILURE in 6m 11s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/1/check/build-openstack-sphinx-docs/2dda57c/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3972ce88daccfc15543474e5997f93a5691f5434","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-10-24 02:19:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/24f8af7/ : FAILURE in 11m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/18552d1/ : TIMED_OUT in 3h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"457206bf767ed1bd81a2ebfe5d245a20a0d834f0","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 16:40:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"be0f155fa4ef33a1f793cdfb9367f7aedd5a23eb","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 16:46:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4c941c87a4ed5b0e860930de116bbc9e8110ec4b","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 16:48:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"42dccd014b26d1b4a511e840d307fe9a9de91cf0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 17:02:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/99f72a5/ : RETRY_LIMIT in 2m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/38dc9f8/ : RETRY_LIMIT in 2m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0c28bcaae76800d16a6543e5197d3514c583322c","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 17:44:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a524c9dedac347061459b5a485684be97ad8e7e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 20:10:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7caf061/ : SUCCESS in 2h 24m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/40b88a0/ : SUCCESS in 2h 23m 00s","accounts_in_message":[],"_revision_number":5},{"id":"6552c5f596a6bf2442a883314aca19ddea7e4819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 00:54:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/5/check/tripleo-ci-centos-7-undercloud-oooq/8ad1779/ : SUCCESS in 1h 15m 13s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/5/check/tripleo-ci-centos-7-undercloud-upgrades/1fb4edb/ : SUCCESS in 1h 16m 55s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/5/check/tripleo-ci-centos-7-containers-multinode/553bf87/ : SUCCESS in 2h 15m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f8a7f78/ : FAILURE in 1h 58m 21s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/5/check/tripleo-buildimage-overcloud-full-centos-7/0548f1d/ : SUCCESS in 30m 55s\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/5/check/openstack-tox-pep8/1e60115/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/5/check/openstack-tox-py27/937bb83/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/5/check/openstack-tox-py35/154b652/ : SUCCESS in 4m 17s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/5/check/openstackclient-check-plugins/26d2bb7/ : SUCCESS in 4m 08s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/5/check/build-openstack-sphinx-docs/a974a55/html/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":5},{"id":"49198f22da91f1003c122662366f6e0232ef6ac1","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-09 16:34:22.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"b13a2a98be6d1ca7fd4049a065db0ef9cd702ec4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 17:45:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/29997f0/ : FAILURE in 1h 08m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b3bda68/ : FAILURE in 1h 09m 23s","accounts_in_message":[],"_revision_number":6},{"id":"dcdae2a8ccdddd2c48ed4cd53a589db7bd69c57c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 21:26:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-undercloud-oooq/2d05620/ : SUCCESS in 1h 19m 25s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-undercloud-upgrades/de6f0d9/ : SUCCESS in 1h 21m 58s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-containers-multinode/3e7580e/ : SUCCESS in 2h 12m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/87fd90e/ : FAILURE in 1h 58m 42s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/6/check/tripleo-buildimage-overcloud-full-centos-7/6e73a0b/ : SUCCESS in 25m 51s\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/6/check/openstack-tox-pep8/4de220f/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/6/check/openstack-tox-py27/3f1eaa7/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/6/check/openstack-tox-py35/96f6147/ : SUCCESS in 3m 59s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/6/check/openstackclient-check-plugins/c6a2bcc/ : SUCCESS in 4m 10s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/6/check/build-openstack-sphinx-docs/34a973f/html/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":6},{"id":"3ea43960f4b4a41bcf552e08100107f35b512659","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-10 14:49:24.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"affc7dea8a9cb8d386023b4d63e6392e2bd4ebf2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 17:26:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/812e401/ : FAILURE in 2h 22m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e4cf2f3/ : SUCCESS in 2h 14m 48s","accounts_in_message":[],"_revision_number":6},{"id":"be3a605e437aec33c030a09592b074644db2f2a4","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-10 17:57:01.000000000","message":"Patch Set 6:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":6},{"id":"f01796d7bea3e050d9e4f10556daabcc11cde926","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:32:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-undercloud-oooq/b6a6e83/ : SUCCESS in 1h 13m 27s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-undercloud-upgrades/2d78fe6/ : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-containers-multinode/12d66c3/ : SUCCESS in 2h 15m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/587305/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0beb0d3/ : FAILURE in 2h 13m 48s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/05/587305/6/check/tripleo-buildimage-overcloud-full-centos-7/59b3817/ : SUCCESS in 28m 23s\n- openstack-tox-pep8 http://logs.openstack.org/05/587305/6/check/openstack-tox-pep8/90291b9/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/05/587305/6/check/openstack-tox-py27/8837646/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/05/587305/6/check/openstack-tox-py35/f1b6aab/ : SUCCESS in 3m 49s\n- openstackclient-check-plugins http://logs.openstack.org/05/587305/6/check/openstackclient-check-plugins/26a0278/ : SUCCESS in 6m 54s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/05/587305/6/check/build-openstack-sphinx-docs/9525ea2/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":6},{"id":"b01789eb40cfe92b63ddce98b270a1d6ed5897b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 20:17:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b97b3ee/ : FAILURE in 2h 19m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/569c065/ : SUCCESS in 2h 12m 25s","accounts_in_message":[],"_revision_number":6},{"id":"d524afb788c3c669448286304206da08d4527a78","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-11 15:21:02.000000000","message":"Patch Set 6:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":6},{"id":"bfbdf4243fb23caa8887e6195ea5c40586bd03cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 18:57:26.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/edd8e4e/ : TIMED_OUT in 3h 35m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1ccb2c6/ : FAILURE in 2h 51m 57s","accounts_in_message":[],"_revision_number":6},{"id":"aab16f977bb4795d722cf6656f3e06264d3d6392","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-12 15:15:48.000000000","message":"Patch Set 6:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":6},{"id":"5b955ae3cdcfa1c9ad8535e3895168dca20e8263","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 17:37:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/478e2d9/ : SUCCESS in 2h 21m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/381a516/ : SUCCESS in 2h 15m 52s","accounts_in_message":[],"_revision_number":6},{"id":"c32a8e40300d2994260dccb2f0da0c22a45a5672","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-15 08:13:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1185d88c28034d8e351cb6fc100d5a9039de3451","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 02:25:40.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Trevor for fixing the merge conflict","accounts_in_message":[],"_revision_number":6},{"id":"5d9499b6dd57c027a28160b9a9f25cbbc0c24de3","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-01 15:38:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c03a5bcb2a116040b6ebd0d0617f5f8272a96e88","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-01 15:49:54.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"c4c46662afa7f91eb84256ace46699946b7b8774","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 18:42:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7941c65/ : SUCCESS in 2h 34m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/587305/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6f5de21/ : SUCCESS in 2h 29m 55s","accounts_in_message":[],"_revision_number":8},{"id":"57b5c8188eda827e2c0abdc0788375cfe85672df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:15:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/05/587305/8/check/tripleo-ci-centos-7-undercloud-oooq/e019df3/ : SUCCESS in 1h 15m 30s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/05/587305/8/check/tripleo-ci-centos-7-undercloud-upgrades/d091316/ : SUCCESS in 1h 27m 56s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/05/587305/8/check/tripleo-ci-centos-7-containers-multinode/9dc24d7/ : SUCCESS in 2h 13m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/05/587305/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4bb9eca/ : FAILURE in 2h 36m 34s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/05/587305/8/check/tripleo-buildimage-overcloud-full-centos-7/3cdee22/ : SUCCESS in 35m 06s\n- openstack-tox-pep8 https://logs.opendev.org/05/587305/8/check/openstack-tox-pep8/4e2f82a/ : SUCCESS in 3m 36s\n- openstack-tox-py27 https://logs.opendev.org/05/587305/8/check/openstack-tox-py27/832481e/ : SUCCESS in 4m 23s\n- openstack-tox-py35 https://logs.opendev.org/05/587305/8/check/openstack-tox-py35/3b4a6ba/ : SUCCESS in 7m 20s\n- openstackclient-check-plugins https://logs.opendev.org/05/587305/8/check/openstackclient-check-plugins/f82234e/ : SUCCESS in 7m 09s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/05/587305/8/check/build-openstack-sphinx-docs/731452e/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":8},{"id":"d2c1ed10c2f84ccbb3a09cf7f33d6de45dbbe882","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-21 18:29:15.000000000","message":"Patch Set 8: Code-Review-1\n\ncan you use git cherry-pick -x so that the commit message is annotated with the commit hash?","accounts_in_message":[],"_revision_number":8},{"id":"0fc344f77c3e673a5b8c317b9306fa3d012522cc","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-21 18:56:15.000000000","message":"Patch Set 8:\n\n\u003e can you use git cherry-pick -x so that the commit message is\n \u003e annotated with the commit hash?\n\nYou want what exactly?  A checkout of stable/queens, and then to run \"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/8 \u0026\u0026 git cherry-pick -x FETCH_HEAD\" or something?  I\u0027m not familiar with the -x argument, what does it provide that isn\u0027t already available?","accounts_in_message":[],"_revision_number":8},{"id":"f62897cafff614b81104441d890db73d98370456","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-29 20:03:03.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"f21a8fe2743fc348538bacceea7d1da6d1284c03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 20:03:30.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"317a65f35483e70b98e523ceac47854b3c1e2696","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 20:14:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"a1741967ca867a0b5dac24c1bbcdf531bf63a6ea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 20:15:54.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"6b09af98fedf08b0323289435e3be6b9fcc33f53","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 20:17:40.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"cf5d40d66435d13e198f66740f385ef91eeea726","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 23:07:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/9e46c8470cf547159502c00eeec0d639 : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/343dc13d49a3485f8c2b79eeca0ed734 : SUCCESS in 1h 56m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1577ab29dd1c420187c029c32187db17 : SUCCESS in 2h 14m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6aea008703ba472888a19d931bc88dc1 : SUCCESS in 2h 49m 58s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/541b01b8f3bf42dca1f8c188671c5bd5 : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd8b04ab67e4cacb271153ff32378e9 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9a46c391ddb47e78527986f1ea9c9d1 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31c05de3d42842858b86a1b39b16f3fc : SUCCESS in 9m 01s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/70151c330ced4a05827368915cee7732 : SUCCESS in 11m 20s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8fa8224d152d49fab338628e3d1a2bf3 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":9},{"id":"1d8830842dac7572130b7654845e52b5d957ec66","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 23:13:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/587305/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cadb99b/ : SUCCESS in 2h 43m 53s","accounts_in_message":[],"_revision_number":9},{"id":"2f1dccb0bae08ca8fa5ad0d7d8f7587c1cfeae67","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-09-03 16:38:33.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"6d6d06636e6b2912ed9825316a0db82653f2d216","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-06 22:08:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0101ed960743a42453b4dd619379d770b6272fa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:08:21.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ee5209968bde8db04d62fa9c3f2845c28f0782e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:30:26.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/fa7cc62c290b47979c4d75a563d4596a : SUCCESS in 1h 20m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8db5e4f5ed954e85b62637df6c7de2dc : SUCCESS in 2h 30m 04s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c508cb120693406ca1e7e7871fcbd735 : SUCCESS in 26m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a251006b2fc149278284c2b97eb1085a : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5493d0f64494eee8e883648a4cc23da : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0fd5bd62f9cd476f85ef3b8bfd149d42 : SUCCESS in 5m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/899b877e85264df288da6d24322d6aa9 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":9},{"id":"005b6171fcf4336a4d95accbe9eb8647d985267b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:30:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7d18de7f325dfea67209f90602c168e7d7ee6c26","revisions":{"94556158ccc792b0277b7105fa93be825b806bf7":{"kind":"REWORK","_number":1,"created":"2018-07-31 07:21:08.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/05/587305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/1"}}},"commit":{"parents":[{"commit":"87d25a3cdec23c1f6e4274489b4f5e8ff9f0860a","subject":"DeployIdentifier should be provided for any parameter actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/87d25a3cdec23c1f6e4274489b4f5e8ff9f0860a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-07-31 07:20:44.000000000","tz":600},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/94556158ccc792b0277b7105fa93be825b806bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/94556158ccc792b0277b7105fa93be825b806bf7"}]},"branch":"refs/heads/stable/queens"},"bacdd5eb758e29e08120b6af456cbadcbdc95112":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-08 16:40:06.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/2"}}},"commit":{"parents":[{"commit":"e717083e2af9d73badef959977ea7608467458eb","subject":"Retry heat stack polling on gateway problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e717083e2af9d73badef959977ea7608467458eb"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 16:40:05.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bacdd5eb758e29e08120b6af456cbadcbdc95112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bacdd5eb758e29e08120b6af456cbadcbdc95112"}]},"branch":"refs/heads/stable/queens"},"c642d2baf97c1972e8f94a081fcbce5f8923cee4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-08 16:46:51.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/3"}}},"commit":{"parents":[{"commit":"9e5ec97754eeb152fb92e5a27904bb8eb1378db4","subject":"Ensure files exist before building/uploading plan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 16:42:35.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c642d2baf97c1972e8f94a081fcbce5f8923cee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c642d2baf97c1972e8f94a081fcbce5f8923cee4"}]},"branch":"refs/heads/stable/queens"},"84f500a6acade0636ec7fa21235ee0ddbb28b6e5":{"kind":"NO_CHANGE","_number":4,"created":"2019-07-08 16:48:53.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/4"}}},"commit":{"parents":[{"commit":"9e5ec97754eeb152fb92e5a27904bb8eb1378db4","subject":"Ensure files exist before building/uploading plan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 16:47:46.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/84f500a6acade0636ec7fa21235ee0ddbb28b6e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/84f500a6acade0636ec7fa21235ee0ddbb28b6e5"}]},"branch":"refs/heads/stable/queens"},"4088418df45cd5a13ff401c42e515386f1321b8a":{"kind":"NO_CHANGE","_number":5,"created":"2019-07-08 17:44:33.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/5"}}},"commit":{"parents":[{"commit":"9e5ec97754eeb152fb92e5a27904bb8eb1378db4","subject":"Ensure files exist before building/uploading plan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 17:44:21.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4088418df45cd5a13ff401c42e515386f1321b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4088418df45cd5a13ff401c42e515386f1321b8a"}]},"branch":"refs/heads/stable/queens"},"e581e803e1befe4d0175203c82a9130fc48dc346":{"kind":"NO_CHANGE","_number":6,"created":"2019-07-09 16:34:22.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/6"}}},"commit":{"parents":[{"commit":"9e5ec97754eeb152fb92e5a27904bb8eb1378db4","subject":"Ensure files exist before building/uploading plan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-09 16:32:40.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e581e803e1befe4d0175203c82a9130fc48dc346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e581e803e1befe4d0175203c82a9130fc48dc346"}]},"branch":"refs/heads/stable/queens"},"01809f4f587e4deba56b2ded49280fddaeac650e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-01 15:38:31.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/7"}}},"commit":{"parents":[{"commit":"de4f45dd2667d3afb1cc55a12b4639e36e30b109","subject":"Allow run_update_ansible_action run with Mistral or Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/de4f45dd2667d3afb1cc55a12b4639e36e30b109"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-08-01 15:36:16.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/01809f4f587e4deba56b2ded49280fddaeac650e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/01809f4f587e4deba56b2ded49280fddaeac650e"}]},"branch":"refs/heads/stable/queens"},"34eb1968ea32ac2aeabb454b338ecd071c99f530":{"kind":"NO_CHANGE","_number":8,"created":"2019-08-01 15:49:54.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/8"}}},"commit":{"parents":[{"commit":"de4f45dd2667d3afb1cc55a12b4639e36e30b109","subject":"Allow run_update_ansible_action run with Mistral or Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/de4f45dd2667d3afb1cc55a12b4639e36e30b109"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-08-01 15:39:10.000000000","tz":-300},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/34eb1968ea32ac2aeabb454b338ecd071c99f530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/34eb1968ea32ac2aeabb454b338ecd071c99f530"}]},"branch":"refs/heads/stable/queens"},"7d18de7f325dfea67209f90602c168e7d7ee6c26":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-08-29 20:03:03.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/05/587305/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/05/587305/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/05/587305/9"}}},"commit":{"parents":[{"commit":"de4f45dd2667d3afb1cc55a12b4639e36e30b109","subject":"Allow run_update_ansible_action run with Mistral or Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/de4f45dd2667d3afb1cc55a12b4639e36e30b109"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-10 05:38:22.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-08-29 20:02:04.000000000","tz":0},"subject":"Allow passing an agent-name as a command line argument","message":"Allow passing an agent-name as a command line argument\n\nAt the same time we deprecate the old AGENT_NAME environment variable in\nfavor of IRONIC_PYTHON_ARGENT_NAME.\n\nBlueprint: multiarch-support\nChange-Id: Ib8c6f944eb2fe492bfc561f543d5b3cd030e9907\n(cherry picked from commit 3cf44eb99483595b3d0dc43dd7994e689398d0f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7d18de7f325dfea67209f90602c168e7d7ee6c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7d18de7f325dfea67209f90602c168e7d7ee6c26"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
