)]}'
{"id":"openstack%2Fpython-tripleoclient~587310","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13","project":"openstack/python-tripleoclient","branch":"stable/queens","topic":"queens/multiarch-support","hashtags":[],"change_id":"I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13","subject":"Actually use the arch and platform to select appropriate images","status":"MERGED","created":"2018-07-31 07:21:08.000000000","updated":"2019-10-16 01:31:06.000000000","submitted":"2019-10-16 01:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"587310-1571189466562-29eb8155","meta_rev_id":"af538622900160ab7c827fc30959f8b5696cbe54","_number":587310,"virtual_id_number":587310,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"date":"2019-10-14 15:23:17.000000000","_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 01:31:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-10 04:00:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"recommended":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-10-11 19:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":2,"date":"2019-09-10 13:55:37.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-15 22:16:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 15:28:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-10 04:00:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 13:55:37.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-10-11 19:34:59.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-10-14 15:23:17.000000000","updated_by":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"reviewer":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"state":"REVIEWER"},{"updated":"2019-10-15 22:16:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-16 01:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5eafdc98598b411816ce4789a7c02a1a03353224","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 07:21:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9acf60ae046992bcced9444092965afbc6ee6e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 07:57:38.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f4d4047fbc130314e2f9e0de4a1bb2951785fd20","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-31 07:57:40.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"056e9e93f460c99b5c0b7610eb353891bdfebf2a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 08:00:19.000000000","message":"Patch Set 1: Workflow-1\n\nInitial backport currently untested","accounts_in_message":[],"_revision_number":1},{"id":"9f03981be853a22569d7ce7340a207199d4471b0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-08-02 23:20:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6c0ed3310816e94dedda11edcfc0005ab8015bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-02 23:41:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens http://logs.rdoproject.org/10/587310/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens/009a1a7/ : FAILURE in 14m 39s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens http://logs.rdoproject.org/10/587310/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens/fb607af/ : FAILURE in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7a9cc2e9e104e958570dc87876de0aef9c8902ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 02:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/1/check/openstack-tox-pep8/1289f72/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/1/check/openstack-tox-py27/c5e439c/ : SUCCESS in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/1/check/openstack-tox-py35/4f326e7/ : SUCCESS in 3m 59s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/1/check/openstackclient-check-plugins/299ecee/ : FAILURE in 4m 31s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/1/check/build-openstack-sphinx-docs/c9505e0/html/ : SUCCESS in 3m 31s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-undercloud-oooq/4d7cc91/ : SUCCESS in 1h 04m 06s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-undercloud-upgrades/ff6b5dd/ : SUCCESS in 1h 21m 10s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-3nodes-multinode/5d10e53/ : SUCCESS in 2h 30m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-containers-multinode/4e23dd2/ : SUCCESS in 2h 10m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/dd79db1/ : TIMED_OUT in 3h 07m 01s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/1/check/tripleo-buildimage-overcloud-full-centos-7/707267e/ : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ba64d28d0b5a73a8a03e77c3415c4566e79f1c9c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-10-23 23:09:04.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"66c8ae6bb136432bb1c8526584483f1a422b6893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 02:06:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-undercloud-oooq/2a5e618/ : SUCCESS in 1h 11m 10s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-undercloud-upgrades/6881733/ : FAILURE in 1h 15m 26s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-containers-multinode/973f287/ : SUCCESS in 2h 25m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/411be21/ : FAILURE in 2h 39m 15s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/1/check/tripleo-buildimage-overcloud-full-centos-7/5474e2d/ : SUCCESS in 30m 46s\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/1/check/openstack-tox-pep8/9895f36/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/1/check/openstack-tox-py27/10fd685/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/1/check/openstack-tox-py35/b080b01/ : SUCCESS in 4m 56s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/1/check/openstackclient-check-plugins/740b043/ : FAILURE in 6m 31s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/1/check/build-openstack-sphinx-docs/8858366/html/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"28507d430477ea066c05595312723cf4441e099c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-10-24 02:12:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/09517ba/ : SUCCESS in 2h 54m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/42f281d/ : FAILURE in 2h 57m 27s","accounts_in_message":[],"_revision_number":1},{"id":"be0fa0074556374b1a7c16d3402091c342320ed1","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 16:48:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"57e450ecf2603d7e25c63a7b79ad01bcaa248fe6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 16:49:20.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"b53d85fc9fe6abc519d970b60252c991ec5f978f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:49:57.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"c320f4a79efb7761046a7b665e0cad853e32cfb1","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 17:44:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"21c29f8426e1b929ab8a2f81eb2caa78186f69ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 17:45:00.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"cc80f6aa592af0e0e09fcc76bd67a035dbb0d1d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:45:07.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"532757f3aa3e9bc54ff4c3892394e2747ebe0627","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 18:20:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"05f847706e503c7391249581f95198f55f351d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 18:21:19.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"aee8b0a35c374ead0441b6ab548e710cb62d8b11","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 18:21:27.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"1cd578bb736f529daf1d42076d0cb4567af38f9a","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-09 16:34:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f467d41be49bc5a7b2bbc68e22b3e98e00784128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:34:40.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"b74fdf4e186272a752e28b49e54274b27d2d820e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 16:34:50.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"0de13b4cb68e1e8ad0de169588bc977a2a6d9579","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-09 17:45:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c6aef875db1d0a3adeb167d3c923cf7606333b81","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 20:20:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/070f3ad/ : FAILURE in 2h 12m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ebc42eb/ : SUCCESS in 2h 18m 38s","accounts_in_message":[],"_revision_number":5},{"id":"a4451ac6a7f660c760a614ce6f284e5324a71a2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 01:41:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/5/check/tripleo-ci-centos-7-undercloud-oooq/daa8b01/ : SUCCESS in 1h 13m 51s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/5/check/tripleo-ci-centos-7-undercloud-upgrades/f564d87/ : SUCCESS in 1h 31m 27s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/5/check/tripleo-ci-centos-7-containers-multinode/bb5c7b8/ : SUCCESS in 2h 18m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/65d7c50/ : FAILURE in 2h 20m 43s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/5/check/tripleo-buildimage-overcloud-full-centos-7/3dbddaa/ : SUCCESS in 33m 53s\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/5/check/openstack-tox-pep8/5c6af59/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/5/check/openstack-tox-py27/b256166/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/5/check/openstack-tox-py35/75f249c/ : SUCCESS in 5m 34s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/5/check/openstackclient-check-plugins/9f3e514/ : SUCCESS in 5m 02s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/5/check/build-openstack-sphinx-docs/3b96313/html/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":5},{"id":"45675d993ef65deeda5c8e281b41d6b32100aa51","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-15 16:28:03.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"38cbdad24c9d98fdd65560af9a20e84921599092","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 17:41:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/784d7e7/ : FAILURE in 1h 10m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1068cdc/ : FAILURE in 1h 11m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c0d55d42b1922ac4d93a646b7b53b95dfd5fd341","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-16 00:37:37.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"baa0d07b26bcdc9b32c7e832fd93303d0c95fb79","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 02:52:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0774c13/ : FAILURE in 2h 10m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1ae0a19/ : SUCCESS in 2h 12m 35s","accounts_in_message":[],"_revision_number":6},{"id":"e3a1ebe03419082a75b706de8969b6d00ab22645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 03:01:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-oooq/91061ef/ : SUCCESS in 1h 16m 24s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-upgrades/e8a7e2d/ : SUCCESS in 1h 19m 25s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-containers-multinode/b01cb78/ : SUCCESS in 2h 19m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5679b51/ : FAILURE in 2h 05m 53s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/6/check/tripleo-buildimage-overcloud-full-centos-7/ff5020a/ : SUCCESS in 33m 52s\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/6/check/openstack-tox-pep8/733fdc3/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/6/check/openstack-tox-py27/a8b5364/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/6/check/openstack-tox-py35/1f766cc/ : FAILURE in 4m 00s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/6/check/openstackclient-check-plugins/574afc8/ : SUCCESS in 6m 00s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/6/check/build-openstack-sphinx-docs/5b90171/html/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":6},{"id":"2a475bec020ed5e8104ce1200241355988ea97fa","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-16 17:04:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e87f418097f8441a0032f0155d1357d0fdf91e14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 19:19:02.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b2d5314/ : FAILURE in 2h 11m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/7cc8503bec274de89193e50e7bb656b2 : RETRY_LIMIT in 6s","accounts_in_message":[],"_revision_number":6},{"id":"ff129a8abd02438bf5157e2ea34728ba163c86dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:18:35.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-oooq/0a5a710/ : SUCCESS in 1h 15m 38s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-upgrades/31da6a1/ : SUCCESS in 1h 33m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-containers-multinode/1b10330/ : FAILURE in 2h 05m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8e64d9f/ : FAILURE in 2h 07m 59s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/6/check/tripleo-buildimage-overcloud-full-centos-7/bc7d87d/ : SUCCESS in 32m 07s\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/6/check/openstack-tox-pep8/765fa61/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/6/check/openstack-tox-py27/5d2933c/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/6/check/openstack-tox-py35/53b8792/ : SUCCESS in 4m 09s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/6/check/openstackclient-check-plugins/73679fb/ : SUCCESS in 4m 50s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/6/check/build-openstack-sphinx-docs/609247e/html/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":6},{"id":"a70bdb7c841427215802322549633f3fe3395979","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-17 15:02:08.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"932306003e5fa359403102ef0c5e4abe005fd37c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 16:17:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bb717d0/ : FAILURE in 1h 07m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fb1b42b/ : FAILURE in 1h 13m 22s","accounts_in_message":[],"_revision_number":6},{"id":"16d9f13dd60ecb3a0730cb903edd64a776383e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:03:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-oooq/6d9ab57/ : SUCCESS in 1h 15m 29s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-undercloud-upgrades/df7a27a/ : SUCCESS in 1h 26m 06s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-containers-multinode/4ba5466/ : FAILURE in 2h 16m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/10/587310/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d2b0c45/ : FAILURE in 2h 13m 49s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/10/587310/6/check/tripleo-buildimage-overcloud-full-centos-7/88fc676/ : FAILURE in 18m 03s\n- openstack-tox-pep8 http://logs.openstack.org/10/587310/6/check/openstack-tox-pep8/1e4a57e/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/10/587310/6/check/openstack-tox-py27/d559240/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/10/587310/6/check/openstack-tox-py35/ea9956a/ : SUCCESS in 3m 52s\n- openstackclient-check-plugins http://logs.openstack.org/10/587310/6/check/openstackclient-check-plugins/a1f214f/ : SUCCESS in 3m 16s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/10/587310/6/check/build-openstack-sphinx-docs/830d546/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"ebfe2882d1c6bcabe570feb8891b32b33f877123","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 03:35:09.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Trevor","accounts_in_message":[],"_revision_number":6},{"id":"5c7f961d009d3477fe5d95a84f4a45b181f01f83","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-01 15:49:54.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"3e4d836326bb9eb0cce232fdf1a8e6c60bc1c882","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 19:37:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8db7b4c/ : TIMED_OUT in 3h 35m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/587310/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fa25948/ : SUCCESS in 2h 25m 01s","accounts_in_message":[],"_revision_number":7},{"id":"d6ec897e2be53d4ebe4d67b067f850cb4c0126d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:03:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/10/587310/7/check/tripleo-ci-centos-7-undercloud-oooq/da6856d/ : SUCCESS in 1h 18m 51s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/10/587310/7/check/tripleo-ci-centos-7-undercloud-upgrades/43fd72f/ : SUCCESS in 1h 31m 13s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/10/587310/7/check/tripleo-ci-centos-7-containers-multinode/36fbf48/ : SUCCESS in 2h 19m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/10/587310/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fbdd806/ : FAILURE in 2h 06m 27s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/10/587310/7/check/tripleo-buildimage-overcloud-full-centos-7/f252e35/ : SUCCESS in 32m 17s\n- openstack-tox-pep8 https://logs.opendev.org/10/587310/7/check/openstack-tox-pep8/75eaeca/ : SUCCESS in 3m 50s\n- openstack-tox-py27 https://logs.opendev.org/10/587310/7/check/openstack-tox-py27/aa174c1/ : SUCCESS in 3m 38s\n- openstack-tox-py35 https://logs.opendev.org/10/587310/7/check/openstack-tox-py35/908c32e/ : SUCCESS in 4m 54s\n- openstackclient-check-plugins https://logs.opendev.org/10/587310/7/check/openstackclient-check-plugins/faf1587/ : SUCCESS in 4m 49s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/10/587310/7/check/build-openstack-sphinx-docs/cbf1dbf/html/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":7},{"id":"11550f14c7560db4df6151526406015828b237c9","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-21 18:29:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"91642724cc7d0f622255bc0456cfffa3e5335902","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-04 15:28:30.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"02873d6182981ea1dd7ea7cded753e0867a1f769","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-09-09 17:15:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a81caa067b91ccbc61f7427390b5ce548cb04e60","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 17:48:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b65db97/ : FAILURE in 31m 23s","accounts_in_message":[],"_revision_number":8},{"id":"ce39dd3928ccb135447906fc5398e595b1a5e5a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:08:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/4dfe2619f97b43f69427b352134f1f90 : FAILURE in 46m 20s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c452d897ac90450aae00a619c84f7fc3 : FAILURE in 1h 24m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cf9b652a747f466395b1591751702a38 : FAILURE in 46m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7d94ee0540df4f75b38635acfb11478c : FAILURE in 32m 45s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/9e2fe11020d44600abb6747262cb315e : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca7f22be9854c9a87afd0f45f7f27fc : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71ae9faf74d044fcbb34bbc245510f6b : FAILURE in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/337b9fe836654288bb65fd0b51f102ad : FAILURE in 4m 25s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/22031b32eb8540d8964deea9f0d46873 : SUCCESS in 4m 59s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce55992200ba4280b934e47058b9a577 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":8},{"id":"f17bb5ef647e6af998ee3b3bcfd9345d7fe4eaab","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-09-10 01:43:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2984a4bdac76d0432484cc208a9df7392617fd8d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-09-10 01:47:03.000000000","message":"Topic changed from queens/multiarch-support-patch7 to queens/multiarch-support","accounts_in_message":[],"_revision_number":9},{"id":"c9539694292051d5fdc5aec543e935341bffc411","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 04:00:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0b67023/ : FAILURE in 2h 15m 08s","accounts_in_message":[],"_revision_number":9},{"id":"af192b5e86b8924237488e022e738d062bd0daba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 04:07:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/57a5fbb4e01744e88ae2d400479a5d53 : SUCCESS in 1h 11m 13s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/30d4625760ba4fdd825f98ad73b10fed : SUCCESS in 1h 25m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4d76cc8ab504441987575b636515ff52 : SUCCESS in 2h 13m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f210dc190abb4d479b6e724fe4d8fbc6 : SUCCESS in 2h 17m 34s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/a6adad17ddbb443a8d2f79baeecb21a0 : SUCCESS in 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18edd01b36ff4315b1a4d91aa019f263 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca00c0e90ea448808215021ffdf3ae75 : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c37c2e7fa7f4db192da442e180320f0 : SUCCESS in 3m 37s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3784f5c7a9ce481e9a2cd553fa3bed0d : SUCCESS in 3m 44s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e65270aff46e4974a63e64221f8cfc5e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":9},{"id":"4ed96a2879bece29ecb65f860bec665c57c18852","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-09-10 13:55:37.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"df286f4b69846b1e3ec519bb2c730eec8781de30","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-09-18 13:32:24.000000000","message":"Patch Set 9:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":9},{"id":"7be84f516754570ce6837a7a86fbe628942aecaf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 16:07:18.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2f67ee3/ : FAILURE in 2h 32m 53s","accounts_in_message":[],"_revision_number":9},{"id":"9fa0d29df302ace0bc9cdf9d9b9999b051807fe5","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-10-11 14:24:18.000000000","message":"Patch Set 9:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":9},{"id":"017f739611f45468b23c8506c162b7a425c28cd4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 16:37:43.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/160045e/ : FAILURE in 1h 58m 21s","accounts_in_message":[],"_revision_number":9},{"id":"c1bda6a30e1910d4337c84782bc087e26524a9ad","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-10-11 19:34:59.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7e092bc40e3b7e0fbc1970becb81288b2f60303a","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-10-14 15:23:17.000000000","message":"Patch Set 9:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":9},{"id":"07d81436f2cd70a9ce530a8755ab840196bcb12d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-14 17:48:21.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/587310/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b1e2860/ : FAILURE in 2h 00m 18s","accounts_in_message":[],"_revision_number":9},{"id":"b8ebff4c2b3bffc34765d325ead364a62c0eaf45","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-15 22:16:55.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f96bd3e00fa76fe3c7686018006538a547bf7970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:17:06.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d0a021676c6ed8a7bcb4eb81f935901e620266df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 01:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"af538622900160ab7c827fc30959f8b5696cbe54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 01:31:06.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/0353980e7e0c42f7aaf80c3b7629a891 : SUCCESS in 2h 12m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3eb2565706564ed29b4de2754c9de982 : SUCCESS in 3h 07m 33s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/9ded5bad08c241f9bc9370597cf6cd61 : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b3d71b4f6b452db10e0fd4f830dd17 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43edf9ab9ed94c668f9a359f3af08993 : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f67575d1fa4e412197ca09558fed9894 : SUCCESS in 4m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a5520828118042f482cd9d874c473ce6 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0e0a3abaffea7cbdc4242913f704dac57a8104e7","revisions":{"d72e5e0be12ef6361d145f836be7613f7e6f87c9":{"kind":"REWORK","_number":1,"created":"2018-07-31 07:21:08.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/10/587310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/1"}}},"commit":{"parents":[{"commit":"c0604ab06ca3148b0726a9ac9463d70ad4e2fa8a","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c0604ab06ca3148b0726a9ac9463d70ad4e2fa8a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-07-31 07:20:44.000000000","tz":600},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: Id82de41e7a49c2d8124fc74125ed51031579aa80\nDepends-On: Idaf05b8efce28cd0cbf339cf693db4f55a693d9b\nDepends-On: I41dce6e25766562db4366021309b8c2b74a8ab80\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d72e5e0be12ef6361d145f836be7613f7e6f87c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d72e5e0be12ef6361d145f836be7613f7e6f87c9"}]},"branch":"refs/heads/stable/queens"},"152469155af7722cd9b2ddf47f9cacd77d1f6841":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-08 16:48:53.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/2"}}},"commit":{"parents":[{"commit":"bb3791b948c0a4f3d9fe1222955c8e1f5cf19141","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bb3791b948c0a4f3d9fe1222955c8e1f5cf19141"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 16:48:25.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: Id82de41e7a49c2d8124fc74125ed51031579aa80\nDepends-On: Idaf05b8efce28cd0cbf339cf693db4f55a693d9b\nDepends-On: I41dce6e25766562db4366021309b8c2b74a8ab80\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/152469155af7722cd9b2ddf47f9cacd77d1f6841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/152469155af7722cd9b2ddf47f9cacd77d1f6841"}]},"branch":"refs/heads/stable/queens"},"ee94d163ed3ee186f3a03c6b4cbd3b6789a17c9b":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-08 17:44:33.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/3"}}},"commit":{"parents":[{"commit":"d647b56ae8d2dd480969a072f22527aa00fe0e7b","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d647b56ae8d2dd480969a072f22527aa00fe0e7b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 17:44:21.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: Id82de41e7a49c2d8124fc74125ed51031579aa80\nDepends-On: Idaf05b8efce28cd0cbf339cf693db4f55a693d9b\nDepends-On: I41dce6e25766562db4366021309b8c2b74a8ab80\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ee94d163ed3ee186f3a03c6b4cbd3b6789a17c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ee94d163ed3ee186f3a03c6b4cbd3b6789a17c9b"}]},"branch":"refs/heads/stable/queens"},"86d5cd0407ae4fa0e6870665c3f2321e991d59f0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-08 18:20:59.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/4"}}},"commit":{"parents":[{"commit":"f8b191ec085c539e56b8d004648a5b4ee8f52b2d","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f8b191ec085c539e56b8d004648a5b4ee8f52b2d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 18:14:46.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: Id82de41e7a49c2d8124fc74125ed51031579aa80\nDepends-On: Idaf05b8efce28cd0cbf339cf693db4f55a693d9b\nDepends-On: I41dce6e25766562db4366021309b8c2b74a8ab80\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/86d5cd0407ae4fa0e6870665c3f2321e991d59f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/86d5cd0407ae4fa0e6870665c3f2321e991d59f0"}]},"branch":"refs/heads/stable/queens"},"6a8120fab14abaa864bd307f115054154b958b25":{"kind":"NO_CHANGE","_number":5,"created":"2019-07-09 16:34:22.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/5"}}},"commit":{"parents":[{"commit":"ca8a4822d2e847145788f67242b30075bf63f707","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ca8a4822d2e847145788f67242b30075bf63f707"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-09 16:32:40.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: Id82de41e7a49c2d8124fc74125ed51031579aa80\nDepends-On: Idaf05b8efce28cd0cbf339cf693db4f55a693d9b\nDepends-On: I41dce6e25766562db4366021309b8c2b74a8ab80\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6a8120fab14abaa864bd307f115054154b958b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6a8120fab14abaa864bd307f115054154b958b25"}]},"branch":"refs/heads/stable/queens"},"b9e5a0ade013e411eaf5ab5de45722022e1f934f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-07-16 00:37:37.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/6"}}},"commit":{"parents":[{"commit":"ca8a4822d2e847145788f67242b30075bf63f707","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ca8a4822d2e847145788f67242b30075bf63f707"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-16 00:35:53.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: https://review.opendev.org/#/c/587329\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b9e5a0ade013e411eaf5ab5de45722022e1f934f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b9e5a0ade013e411eaf5ab5de45722022e1f934f"}]},"branch":"refs/heads/stable/queens"},"f72381dd42e7a4f9beed47324df9bc44943d7cec":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-01 15:49:54.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/7"}}},"commit":{"parents":[{"commit":"19aca2302a25ed1d445531442e0cccd1cfca795b","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/19aca2302a25ed1d445531442e0cccd1cfca795b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-08-01 15:45:38.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: https://review.opendev.org/#/c/587329\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f72381dd42e7a4f9beed47324df9bc44943d7cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f72381dd42e7a4f9beed47324df9bc44943d7cec"}]},"branch":"refs/heads/stable/queens"},"9f3c8784d31fdf415d6c8050b4dfc5559375e7fb":{"kind":"REWORK","_number":8,"created":"2019-09-09 17:15:45.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/8"}}},"commit":{"parents":[{"commit":"c0c24654461c48e3744c9ecbd03e14b8cf9aa45c","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c0c24654461c48e3744c9ecbd03e14b8cf9aa45c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-09-09 17:15:12.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: https://review.opendev.org/#/c/587329\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9f3c8784d31fdf415d6c8050b4dfc5559375e7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9f3c8784d31fdf415d6c8050b4dfc5559375e7fb"}]},"branch":"refs/heads/stable/queens"},"0e0a3abaffea7cbdc4242913f704dac57a8104e7":{"kind":"REWORK","_number":9,"created":"2019-09-10 01:43:55.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/10/587310/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/10/587310/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/10/587310/9"}}},"commit":{"parents":[{"commit":"2fae8d9cbb788cf73558d00fc7e35c7daa224df6","subject":"Add --platform support when uploading images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2fae8d9cbb788cf73558d00fc7e35c7daa224df6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-12 06:27:05.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-09-10 01:40:50.000000000","tz":-300},"subject":"Actually use the arch and platform to select appropriate images","message":"Actually use the arch and platform to select appropriate images\n\nThe prior changes in this series enabled uploading of images tagged in\nsuch a way that they could be used with multiple architectures, this\nchange uses the data from instckenv.json to select appropriate deploy\nimages for each node as it\u0027s registered to ironic.\n\nBlueprint: multiarch-support\nDepends-On: https://review.opendev.org/#/c/587329\nChange-Id: I7c84f3035853d8ee7b8d45895e7acce8e9dd3d13\n(cherry picked from commit 560b79cb1b1739747f6a81ace390b63d63cdf3c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0e0a3abaffea7cbdc4242913f704dac57a8104e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0e0a3abaffea7cbdc4242913f704dac57a8104e7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
