)]}'
{"id":"openstack%2Fpython-tripleoclient~587311","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~I996248f06ff47a7e20398848ae047eb4c2227dec","project":"openstack/python-tripleoclient","branch":"stable/queens","topic":"queens/multiarch-support","hashtags":[],"change_id":"I996248f06ff47a7e20398848ae047eb4c2227dec","subject":"Switch to using assert_has_calls() instead of assertEqual()","status":"MERGED","created":"2018-07-31 07:21:08.000000000","updated":"2019-10-16 01:31:07.000000000","submitted":"2019-10-16 01:31:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"587311-1571189467572-0f5a3271","meta_rev_id":"8b697fc43816140cdf6a5033c5cbc4abd4d7a74e","_number":587311,"virtual_id_number":587311,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 01:31:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-10 04:04:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"recommended":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-09-10 01:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-09-10 01:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":2,"date":"2019-09-10 01:43:55.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-09-10 01:43:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-15 22:17:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 15:02:15.000000000","updated_by":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"reviewer":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"state":"REVIEWER"},{"updated":"2019-07-17 17:55:56.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-21 18:30:52.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-09-04 15:28:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-10 04:04:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-15 22:17:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-16 01:31:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb4cfd412f1e640d4b09df58ba89a88b3b0e700b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 07:21:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"991b37ac45958acb0fd7635f16097b78d15305dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 07:57:38.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6eba1df171fa265d3e0af144a86e42f67c21caa5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-31 07:57:40.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e2101e242d473bf0fe65107ad740a9120d44b5fb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-07-31 08:00:18.000000000","message":"Patch Set 1: Workflow-1\n\nInitial backport currently untested","accounts_in_message":[],"_revision_number":1},{"id":"282af24f3a9db6260bab9bf5b4ff80040757fc43","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-08-02 23:20:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"daf4735aae2f024ce69a64315b72389f7f037e1e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-02 23:56:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens http://logs.rdoproject.org/11/587311/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens/94b94a7/ : FAILURE in 27m 47s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens http://logs.rdoproject.org/11/587311/1/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-queens/7a462d8/ : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4e0ef1062e433750b2f8a6f5eecf62e6f7b793ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 02:32:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/587311/1/check/openstack-tox-pep8/5c27d54/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/11/587311/1/check/openstack-tox-py27/0c928e5/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/11/587311/1/check/openstack-tox-py35/4c9ae0b/ : SUCCESS in 4m 03s\n- openstackclient-check-plugins http://logs.openstack.org/11/587311/1/check/openstackclient-check-plugins/3cd91d4/ : FAILURE in 6m 53s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/11/587311/1/check/build-openstack-sphinx-docs/12d4b99/html/ : SUCCESS in 3m 09s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-undercloud-oooq/2aea44c/ : SUCCESS in 1h 09m 27s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-undercloud-upgrades/b9185ae/ : SUCCESS in 59m 50s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-3nodes-multinode/d90db26/ : FAILURE in 1h 11m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-containers-multinode/0b86503/ : SUCCESS in 2h 14m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ee54ec9/ : TIMED_OUT in 3h 07m 03s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/587311/1/check/tripleo-buildimage-overcloud-full-centos-7/b15b038/ : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"69ee01da3147d5ac553d2a58b5365e271e631909","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-10-23 23:08:57.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6078733a5fb1acfb4661b2c64f9c699063fa4b61","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-10-24 01:33:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bf96865/ : FAILURE in 2h 21m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/587311/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1b290b8/ : FAILURE in 2h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"dafb18baaf5b7b5ecf38acf3496c0e3995226846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 01:54:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-undercloud-oooq/21a8da1/ : SUCCESS in 1h 16m 00s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-undercloud-upgrades/67e12aa/ : FAILURE in 1h 17m 47s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-containers-multinode/cd6cfe9/ : SUCCESS in 2h 40m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/587311/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0838238/ : FAILURE in 2h 36m 18s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/587311/1/check/tripleo-buildimage-overcloud-full-centos-7/5afafec/ : SUCCESS in 31m 25s\n- openstack-tox-pep8 http://logs.openstack.org/11/587311/1/check/openstack-tox-pep8/de439ed/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/11/587311/1/check/openstack-tox-py27/eb1f23c/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/11/587311/1/check/openstack-tox-py35/14c457b/ : SUCCESS in 5m 24s\n- openstackclient-check-plugins http://logs.openstack.org/11/587311/1/check/openstackclient-check-plugins/100cb32/ : FAILURE in 5m 26s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/11/587311/1/check/build-openstack-sphinx-docs/91b88a5/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"aad13af337af4587c19e487e354be9f3ba7d2c70","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 16:48:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8c75b8a4d72e058ab37ea6db06f1aa8c47aa1604","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 16:49:25.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"4920a67f2b62dbcc43a3e817f4350561e3f0a443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:49:57.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"17580623f7813e5d8e358ac3cd2491085b7f99fa","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 17:44:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6fd0b9d071e7ae5ca92f988e5fe30676c1fb70e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:44:59.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"1e2dc088d545763d7adffc16676524a0e335a6c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 17:45:01.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"07d111f6cdfae611744bf7d662b283ab2e716db7","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-08 18:20:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"43b060d96c6dedc69904e9d56cf77145ba81cf68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 18:21:23.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"e4eb46b3a22e85462c9741e8263df8c1042c3875","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 18:21:28.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"22eb0afd7b5bb6098bc24bc07137657dee7cc30e","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-09 16:34:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6c445e1b2c4fb049ab59849cbdce554d58efd85a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:34:42.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"d123d3cb4a48f008965a854e0fc4b923ad6e02fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 16:34:54.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"489a89b14a1f19fed0186421deca499aa7d24de1","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-16 00:37:37.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a9a36a501d507db1422fc1c99cb39370cbcb630f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 02:51:14.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0d29a5e/ : FAILURE in 2h 11m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/587311/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1724d77/ : SUCCESS in 2h 05m 57s","accounts_in_message":[],"_revision_number":6},{"id":"7fc031edcd3c82ec160923cc81abdc3117d6c410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 03:05:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-undercloud-oooq/7e7dd6d/ : SUCCESS in 1h 17m 32s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-undercloud-upgrades/3485819/ : SUCCESS in 1h 28m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-containers-multinode/e9d9030/ : SUCCESS in 2h 24m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ee6af38/ : FAILURE in 1h 48m 13s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/587311/6/check/tripleo-buildimage-overcloud-full-centos-7/6d806c5/ : SUCCESS in 27m 41s\n- openstack-tox-pep8 http://logs.openstack.org/11/587311/6/check/openstack-tox-pep8/d705bdd/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/11/587311/6/check/openstack-tox-py27/70aa9e3/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/11/587311/6/check/openstack-tox-py35/cbeb57a/ : SUCCESS in 3m 39s\n- openstackclient-check-plugins http://logs.openstack.org/11/587311/6/check/openstackclient-check-plugins/8858fba/ : SUCCESS in 4m 14s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/11/587311/6/check/build-openstack-sphinx-docs/bd49677/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"7956c1f6956d4a75ac0a948f4250da3564cbcb76","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-07-17 15:02:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b6bac7c1d69c684a0aa0eadbe1b0699dd05773d2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 17:19:07.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b9801ea/ : FAILURE in 1h 10m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/587311/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1a2125b/ : SUCCESS in 2h 14m 51s","accounts_in_message":[],"_revision_number":6},{"id":"6d5192b2a2ca9fba7230b8c51bf265b323f08901","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-17 17:55:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7e84bb0d1b3b3adaa956f0fc2bcd5664672c7e6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:25:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-undercloud-oooq/448ec92/ : SUCCESS in 1h 21m 42s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-undercloud-upgrades/bd048fb/ : SUCCESS in 1h 25m 38s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-containers-multinode/12ab5fd/ : FAILURE in 2h 23m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/587311/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7d04e2a/ : FAILURE in 2h 28m 16s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/587311/6/check/tripleo-buildimage-overcloud-full-centos-7/247e6d3/ : FAILURE in 19m 41s\n- openstack-tox-pep8 http://logs.openstack.org/11/587311/6/check/openstack-tox-pep8/e02d771/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/11/587311/6/check/openstack-tox-py27/668e37a/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/11/587311/6/check/openstack-tox-py35/039428b/ : SUCCESS in 4m 25s\n- openstackclient-check-plugins http://logs.openstack.org/11/587311/6/check/openstackclient-check-plugins/07675e3/ : SUCCESS in 4m 05s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/11/587311/6/check/build-openstack-sphinx-docs/b5d2160/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"afb21214deab9c1577b7335a93042369430457cc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-07-19 03:36:15.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks Trevor","accounts_in_message":[],"_revision_number":6},{"id":"b243d5101d7b0a07f6866181c80d5817de083675","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-08-01 15:49:54.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"9bb9cfcd20a8fc7d4adb6cd83c4016d5e77aa684","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 18:25:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/269382b/ : FAILURE in 2h 31m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/587311/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5494296/ : SUCCESS in 2h 30m 22s","accounts_in_message":[],"_revision_number":7},{"id":"eacbd1fed9072b0012fe2ca490000fbc286bdada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 02:19:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/11/587311/7/check/tripleo-ci-centos-7-undercloud-oooq/6f9c7ba/ : SUCCESS in 1h 19m 39s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/11/587311/7/check/tripleo-ci-centos-7-undercloud-upgrades/60c7f28/ : SUCCESS in 1h 20m 27s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/11/587311/7/check/tripleo-ci-centos-7-containers-multinode/32a8302/ : SUCCESS in 2h 23m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/11/587311/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f3a8725/ : FAILURE in 2h 26m 40s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/11/587311/7/check/tripleo-buildimage-overcloud-full-centos-7/1d6ecbf/ : SUCCESS in 31m 06s\n- openstack-tox-pep8 https://logs.opendev.org/11/587311/7/check/openstack-tox-pep8/afac07c/ : SUCCESS in 4m 44s\n- openstack-tox-py27 https://logs.opendev.org/11/587311/7/check/openstack-tox-py27/1c9b390/ : SUCCESS in 4m 31s\n- openstack-tox-py35 https://logs.opendev.org/11/587311/7/check/openstack-tox-py35/caba3d2/ : SUCCESS in 4m 58s\n- openstackclient-check-plugins https://logs.opendev.org/11/587311/7/check/openstackclient-check-plugins/56b842a/ : SUCCESS in 6m 33s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/11/587311/7/check/build-openstack-sphinx-docs/560d589/html/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":7},{"id":"b012fa2ea5cb99b5b0ac6d90c1476357ab6f3014","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-21 18:30:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2d57e8394196f54b5230012a42f0e5ff9b88a832","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-04 15:28:49.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a7482ef52196e9a3cd469783af00067a22d2d3d9","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-09-09 17:15:45.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ec4db9e6a080919964587baa323fd32c1c179856","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 17:50:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1ba8dec/ : FAILURE in 32m 13s","accounts_in_message":[],"_revision_number":8},{"id":"2e0dd3fcdb017b819fa68bfeb298380d41da8539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:02:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/c5ee3fa623d742448c062550a048c51a : FAILURE in 24m 46s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1d329ebf7d7c4a239dd658258e677589 : FAILURE in 1h 05m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4548c9e2f12a42a7983825efaaac4395 : FAILURE in 45m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ecd5bae8fa1b46c7a16903cb32b9fc64 : FAILURE in 33m 03s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/88526dc3d19e4ed79e4faf5824e89691 : SUCCESS in 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9ca393093e421692f8f605512fdb84 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/440964a0046e4f33b9bf01845e069b8b : FAILURE in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4cfb8bfd80624cb59d63e8804d718f19 : FAILURE in 4m 48s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4f722c11fc3d47049b3e1544a28ba172 : SUCCESS in 4m 28s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/55aa6f9822e44926be55c99e7b979fc2 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":8},{"id":"0941c6b9bbee11fca1bd23d67d78e8c68050c1ee","author":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"date":"2019-09-10 01:43:55.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"7dea05b9c32b99930b955c3b50c14f0f60779a80","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-09-10 01:47:10.000000000","message":"Topic changed from queens/multiarch-support-patch7 to queens/multiarch-support","accounts_in_message":[],"_revision_number":9},{"id":"da23b40f486380e4b18f2eaf92f46ae2df1dec15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 04:04:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/587311/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fc430e2/ : FAILURE in 2h 18m 32s","accounts_in_message":[],"_revision_number":9},{"id":"0e77cbbe83dc1011df4a3af16f4fd74bc1810b57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 04:33:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/48199eb3f1954dbb9be1c0770244275b : SUCCESS in 1h 20m 50s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b63d1e8408e14641a055e9ad1694ab2e : SUCCESS in 1h 25m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4ad791f3925f4b658326674c231bef97 : SUCCESS in 2h 23m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ba4085e381fe4565996d84f1c8896ca3 : SUCCESS in 2h 47m 41s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/a2e837ce7b794873bb320a554c5a6257 : SUCCESS in 24m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3490477a3db846d3a064d00f1d17fb16 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e9f5ad653264edc8ab9f3724b418abc : SUCCESS in 3m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9bc91b35edeb48498195db2eed5bf3c7 : SUCCESS in 4m 08s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/81188a8ef4f640c4a81815ea7b9669f2 : SUCCESS in 3m 43s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae7cac61989a4d1d888500e22202a152 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":9},{"id":"4f066a260f76b0bf1a350d8914bb37e810ebe979","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-15 22:17:04.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"482aee2acf069e418a33a2892b5e294ae0a6580a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:17:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5ed7a6a884bebeac6ecb2d592db7dc282df26dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 01:31:07.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/84623aa2ad4a4e9295623fdd270a9b33 : SUCCESS in 2h 03m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c580e6d65c3941cf83217aca0c82b132 : SUCCESS in 2h 19m 37s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/7e495a3359bc4b0ba27be5f042dfaa4d : SUCCESS in 31m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ffc709e7db4b9394afa5f47873139b : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73a8a8b38134472cb8820f906fb1390f : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/63bc006208154d02a1bad52a9cab9ba1 : SUCCESS in 4m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a8baf68987a40a28f2806ba7568a243 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":9},{"id":"b214fec748978c4dc1172f1209c25cbea69ea19b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 01:31:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3c3d813cf523db0b96bae09d8f502df8f9beaa20","revisions":{"806ed89f06dbaf1cceb506498f04aa662b179ec9":{"kind":"REWORK","_number":1,"created":"2018-07-31 07:21:08.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/11/587311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/1"}}},"commit":{"parents":[{"commit":"d72e5e0be12ef6361d145f836be7613f7e6f87c9","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d72e5e0be12ef6361d145f836be7613f7e6f87c9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-07-31 07:20:44.000000000","tz":600},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/806ed89f06dbaf1cceb506498f04aa662b179ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/806ed89f06dbaf1cceb506498f04aa662b179ec9"}]},"branch":"refs/heads/stable/queens"},"22b066cd666e9f1cab263523561a8bcfe8538828":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-08 16:48:53.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/2"}}},"commit":{"parents":[{"commit":"152469155af7722cd9b2ddf47f9cacd77d1f6841","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/152469155af7722cd9b2ddf47f9cacd77d1f6841"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 16:48:25.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/22b066cd666e9f1cab263523561a8bcfe8538828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/22b066cd666e9f1cab263523561a8bcfe8538828"}]},"branch":"refs/heads/stable/queens"},"52882d5ba803efc22be6ac67d5607fbd3f338f4f":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-08 17:44:33.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/3"}}},"commit":{"parents":[{"commit":"ee94d163ed3ee186f3a03c6b4cbd3b6789a17c9b","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ee94d163ed3ee186f3a03c6b4cbd3b6789a17c9b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 17:44:21.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/52882d5ba803efc22be6ac67d5607fbd3f338f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/52882d5ba803efc22be6ac67d5607fbd3f338f4f"}]},"branch":"refs/heads/stable/queens"},"579601b8d2777d72893201a92331f787973984d6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-08 18:20:59.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/4"}}},"commit":{"parents":[{"commit":"86d5cd0407ae4fa0e6870665c3f2321e991d59f0","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/86d5cd0407ae4fa0e6870665c3f2321e991d59f0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-08 18:15:00.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/579601b8d2777d72893201a92331f787973984d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/579601b8d2777d72893201a92331f787973984d6"}]},"branch":"refs/heads/stable/queens"},"1c19dfa34e1bdbb34202529b2d4b234d6a3db6df":{"kind":"NO_CHANGE","_number":5,"created":"2019-07-09 16:34:22.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/5"}}},"commit":{"parents":[{"commit":"6a8120fab14abaa864bd307f115054154b958b25","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6a8120fab14abaa864bd307f115054154b958b25"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-09 16:32:40.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1c19dfa34e1bdbb34202529b2d4b234d6a3db6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1c19dfa34e1bdbb34202529b2d4b234d6a3db6df"}]},"branch":"refs/heads/stable/queens"},"5d93b4d37272aad6da22bae60849f1e8dc067ea9":{"kind":"NO_CHANGE","_number":6,"created":"2019-07-16 00:37:37.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/6"}}},"commit":{"parents":[{"commit":"b9e5a0ade013e411eaf5ab5de45722022e1f934f","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b9e5a0ade013e411eaf5ab5de45722022e1f934f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-07-16 00:37:24.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d93b4d37272aad6da22bae60849f1e8dc067ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d93b4d37272aad6da22bae60849f1e8dc067ea9"}]},"branch":"refs/heads/stable/queens"},"62a0d3b114ff5dcd289d5a5dd02800cf5274ba9a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-01 15:49:54.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/7"}}},"commit":{"parents":[{"commit":"f72381dd42e7a4f9beed47324df9bc44943d7cec","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f72381dd42e7a4f9beed47324df9bc44943d7cec"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-08-01 15:45:38.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/62a0d3b114ff5dcd289d5a5dd02800cf5274ba9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/62a0d3b114ff5dcd289d5a5dd02800cf5274ba9a"}]},"branch":"refs/heads/stable/queens"},"2e61b2a5718a4115ec37f40bfa71ad35a276ff94":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-09 17:15:45.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/8"}}},"commit":{"parents":[{"commit":"9f3c8784d31fdf415d6c8050b4dfc5559375e7fb","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9f3c8784d31fdf415d6c8050b4dfc5559375e7fb"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-09-09 17:15:12.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2e61b2a5718a4115ec37f40bfa71ad35a276ff94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2e61b2a5718a4115ec37f40bfa71ad35a276ff94"}]},"branch":"refs/heads/stable/queens"},"3c3d813cf523db0b96bae09d8f502df8f9beaa20":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-10 01:43:55.000000000","uploader":{"_account_id":10806,"name":"Trevor Vardeman","email":"tvardema@redhat.com","username":"tvardeman"},"ref":"refs/changes/11/587311/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/587311/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/587311/9"}}},"commit":{"parents":[{"commit":"0e0a3abaffea7cbdc4242913f704dac57a8104e7","subject":"Actually use the arch and platform to select appropriate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0e0a3abaffea7cbdc4242913f704dac57a8104e7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2018-01-19 05:23:26.000000000","tz":660},"committer":{"name":"Trevor Vardeman","email":"tvardema@redhat.com","date":"2019-09-10 01:40:56.000000000","tz":-300},"subject":"Switch to using assert_has_calls() instead of assertEqual()","message":"Switch to using assert_has_calls() instead of assertEqual()\n\nWhile adding multiarch support I noticed that some tests I was copying\nhad switched the values of actual and reference.  While this doesn\u0027t\nalter the correctness of the tests it is somewhat strange when the tests\nfail and a human is looking at the output.\n\nAlso during review Steve Baker noted we could use assert_has_calls() so\navoid confusing humans by switching :)\n\nChange-Id: I996248f06ff47a7e20398848ae047eb4c2227dec\n(cherry picked from commit 351633e4a396e38accdbcc99db1e82ba36f2f537)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3c3d813cf523db0b96bae09d8f502df8f9beaa20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3c3d813cf523db0b96bae09d8f502df8f9beaa20"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
