)]}'
{"id":"openstack%2Fpython-tripleoclient~636849","triplet_id":"openstack%2Fpython-tripleoclient~master~I427887467678c11c74382b6805e32d1455641294","project":"openstack/python-tripleoclient","branch":"master","topic":"use_www_authenticate_uri","hashtags":[],"change_id":"I427887467678c11c74382b6805e32d1455641294","subject":"Bump keystone middleware version to 4.18","status":"MERGED","created":"2019-02-14 07:56:46.000000000","updated":"2019-06-26 22:51:07.000000000","submitted":"2019-06-26 22:51:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636849-1561589467792-6b32c878","meta_rev_id":"b9a06a628818e86d9ab7527fdf5bd5c6aac23d13","_number":636849,"virtual_id_number":636849,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-06-26 22:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 22:00:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-20 19:23:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-20 19:23:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-06-20 19:23:23.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-06-20 19:23:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-20 19:23:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-06-21 02:44:20.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 09:39:23.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-06-07 13:17:44.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-06-07 19:21:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-20 22:00:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-21 02:44:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-26 22:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a4401ddb76b5c32fbb9ba37d6d1545942b263ad","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-02-14 07:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"636ef33eb9a4e4c6874bc475487ef015bdbc9a8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-14 08:12:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/636849/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7878d88/ : FAILURE in 13m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/636849/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0323a0a/ : FAILURE in 13m 57s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/699b8d4/ : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"06327ae9132b7269c9d25ca1d424f4c7551587d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 09:49:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/636849/1/check/requirements-check/a50df6b/ : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/1/check/openstack-tox-lower-constraints/9aa671f/ : SUCCESS in 4m 50s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/1/check/openstackclient-check-plugins/8483292/ : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/1/check/openstack-tox-pep8/03bd109/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/1/check/openstack-tox-py27/fa5466c/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/1/check/openstack-tox-py35/0646d4a/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/1/check/openstack-tox-py36/f6baa1b/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/1/check/openstack-tox-py37/0296036/ : SUCCESS in 3m 25s\n- openstack-tox-docs http://logs.openstack.org/49/636849/1/check/openstack-tox-docs/a805366/html/ : SUCCESS in 3m 08s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/1/check/tripleo-buildimage-overcloud-full-centos-7/780d9b1/ : FAILURE in 13m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-containers-multinode/9c1618c/ : FAILURE in 24m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-undercloud-containers/ffb7f23/ : FAILURE in 17m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bd3429b/ : FAILURE in 19m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/97bde44/ : FAILURE in 16m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-standalone/21d5ad7/ : FAILURE in 17m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/1/check/tripleo-ci-fedora-28-standalone/99d4f59/ : SUCCESS in 1h 47m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-standalone-upgrade/fe8c5ea/ : FAILURE in 17m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a096ef5/ : FAILURE in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"331ea293556b7254a69693427713fdb9bab89dcc","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-03-11 19:27:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3198b91d041a2801066268e4005436dcbd8ea6ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 00:03:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/636849/1/check/requirements-check/5994d4b/ : SUCCESS in 2m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/1/check/openstack-tox-lower-constraints/ede8e09/ : SUCCESS in 5m 20s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/1/check/openstackclient-check-plugins/2eb5d59/ : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/1/check/openstack-tox-pep8/61562f5/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/1/check/openstack-tox-py27/ef268bd/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/1/check/openstack-tox-py35/4530a1d/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/1/check/openstack-tox-py36/b9e47d3/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/1/check/openstack-tox-py37/cd24e94/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/49/636849/1/check/openstack-tox-docs/068c4a1/html/ : SUCCESS in 3m 46s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/1/check/tripleo-buildimage-overcloud-full-centos-7/45ae85b/ : SUCCESS in 17m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/1/check/tripleo-build-containers-centos-7/1140730/ : SUCCESS in 38m 02s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/1/check/tripleo-build-containers-centos-7-buildah/a35fbb3/ : SUCCESS in 52m 18s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-containers-multinode/30e17b4/ : SUCCESS in 2h 13m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-undercloud-containers/3b8d95d/ : SUCCESS in 1h 42m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/faccfba/ : SUCCESS in 2h 35m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/383a4c7/ : SUCCESS in 2h 35m 48s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-standalone/cda0552/ : SUCCESS in 1h 54m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/1/check/tripleo-ci-fedora-28-standalone/2d025e1/ : FAILURE in 26m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-standalone-upgrade/4dc5d86/ : FAILURE in 1h 55m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/573b911/ : SUCCESS in 2h 09m 11s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/c64921f/ : SUCCESS in 2h 29m 20s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/08dd12c/ : FAILURE in 1h 35m 22s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/49/636849/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/b0bcdb1/ : FAILURE in 1h 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e05fa7ac95dd6964d3399a0cc6f70d97e94c52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 02:10:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/ac77693524ce4c1db9002c96f601dc43 : POST_FAILURE in 4h 01m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze03.softwarefactory-project.io:7979/99ab43e02981482fb9f47c13eec208b7 : TIMED_OUT in 4h 03m 08s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/56cdb40/ : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d8c1d48fe0fe96e72e4edd8a75ebd18e9774e002","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-03-27 16:55:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9965d48e2bd1f759bff2fb15b9c31f114f149cf","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-04-17 09:38:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55e36e38c41a59c507d7b6b854ad3303016ac5c7","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-04-17 09:39:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df58f7e6130ca2b5bf94eff130a38c5895ac8722","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-06-07 13:17:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a7c6e13ff6ab981da6352efc6314db8401e56ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:18:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5b51f4f762a1a817f727ca2f38034c4b1081e1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:46:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/636849/1/gate/requirements-check/675bf7d/ : FAILURE in 2m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/1/gate/openstack-tox-lower-constraints/0ebb376/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/1/gate/openstack-tox-pep8/6e3ada9/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/1/gate/openstack-tox-py27/bf01ef3/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/1/gate/openstack-tox-py35/c617e7a/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/1/gate/openstack-tox-py36/57c6a8e/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/1/gate/openstack-tox-py37/a20ed26/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/49/636849/1/gate/openstack-tox-docs/655a0b2/html/ : SUCCESS in 3m 09s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/1/gate/tripleo-buildimage-overcloud-full-centos-7/ed6bbae/ : SUCCESS in 22m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/1/gate/tripleo-build-containers-centos-7/37c4530/ : SUCCESS in 50m 22s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/49/636849/1/gate/tripleo-build-containers-centos-7-stein/bf47038/ : SUCCESS in 44m 24s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/49/636849/1/gate/tripleo-build-containers-centos-7-rocky/1feccd6/ : SUCCESS in 38m 02s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/1/gate/tripleo-build-containers-centos-7-buildah/23fff9a/ : SUCCESS in 1h 02m 29s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/49/636849/1/gate/tripleo-build-containers-centos-7-buildah-stein/224c012/ : SUCCESS in 57m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/49/636849/1/gate/tripleo-ci-centos-7-containers-multinode/83b12c7/ : SUCCESS in 2h 04m 49s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/1/gate/tripleo-ci-centos-7-standalone/947411a/ : FAILURE in 2h 14m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/1/gate/tripleo-ci-fedora-28-standalone/71cf284/ : SUCCESS in 2h 15m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/49/636849/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4a16ec3/ : SUCCESS in 2h 14m 50s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/1/gate/tripleo-ci-centos-7-standalone-upgrade-stein/0d3058b/ : SUCCESS in 1h 58m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/636849/1/gate/tripleo-ci-centos-7-undercloud-containers/fb3708f/ : SUCCESS in 1h 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3076a1cf9f464a067793512650b87a0375d3fe11","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 19:20:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a64f7012820598177be18caf6c6288a5e7b65547","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 19:21:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a10da1ee47dbf2cfd4778234ce8991c090aac8ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 22:04:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/636849/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/34d1fb1/ : SUCCESS in 2h 26m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/636849/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/945d278/ : SUCCESS in 2h 41m 43s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/231d869/ : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0654139d20c90e5e633f46bd1d01d5a37314fd1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 07:24:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/636849/2/check/requirements-check/57a1f5e/ : FAILURE in 2m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/2/check/openstack-tox-lower-constraints/9a2b8fc/ : SUCCESS in 4m 50s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/2/check/openstackclient-check-plugins/92610cb/ : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/2/check/openstack-tox-pep8/ba47aa1/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/2/check/openstack-tox-py27/ed5c8fb/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/2/check/openstack-tox-py35/cf5dcce/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/2/check/openstack-tox-py36/45b835b/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/2/check/openstack-tox-py37/aea156e/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/49/636849/2/check/openstack-tox-docs/74cba00/html/ : SUCCESS in 3m 36s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/2/check/tripleo-buildimage-overcloud-full-centos-7/637cf86/ : SUCCESS in 20m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/2/check/tripleo-build-containers-centos-7/58324c9/ : SUCCESS in 49m 34s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/49/636849/2/check/tripleo-build-containers-centos-7-stein/e88a89d/ : SUCCESS in 53m 13s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/49/636849/2/check/tripleo-build-containers-centos-7-rocky/053bf1c/ : SUCCESS in 48m 35s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/2/check/tripleo-build-containers-centos-7-buildah/d5b6351/ : SUCCESS in 45m 06s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/49/636849/2/check/tripleo-build-containers-centos-7-buildah-stein/32fa891/ : SUCCESS in 59m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-containers-multinode/c532e47/ : SUCCESS in 2h 05m 17s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-undercloud-containers/f38b8e0/ : SUCCESS in 1h 30m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d3f1737/ : SUCCESS in 2h 08m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/af6fc6c/ : SUCCESS in 2h 39m 07s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-standalone/c1e7c06/ : SUCCESS in 1h 55m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/2/check/tripleo-ci-fedora-28-standalone/0a1a944/ : SUCCESS in 1h 55m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-standalone-upgrade/70c0878/ : SUCCESS in 2h 05m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/e55f5eb/ : SUCCESS in 1h 58m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ed57f6f/ : FAILURE in 1h 18m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8bbfb6e/ : SUCCESS in 2h 07m 52s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/a4aedf9/ : FAILURE in 1h 28m 38s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/49/636849/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/7857797/ : SUCCESS in 2h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e86d54cba7bdc5e97d7595b7c323090def593d04","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-19 21:01:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ef14f651cd8e91d0da6875eb1a492dc1ec7a25e4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-19 21:05:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a8d2f88f1719aa47d2d5b7c10dc65ca0da88266e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 23:40:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/636849/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b26037e/ : SUCCESS in 2h 26m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/636849/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9ac460c/ : SUCCESS in 2h 33m 57s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ffdcd60/ : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":4},{"id":"b41070736d0756618bd203f9f568fa75a7e2817c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 01:26:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/636849/4/check/requirements-check/c8ea3c2/ : FAILURE in 2m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/4/check/openstack-tox-lower-constraints/7d46ab4/ : SUCCESS in 6m 47s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/4/check/openstackclient-check-plugins/c9dced5/ : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/4/check/openstack-tox-pep8/c1182a2/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/4/check/openstack-tox-py27/a18cccc/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/4/check/openstack-tox-py35/1210da8/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/4/check/openstack-tox-py36/6d22c6d/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/4/check/openstack-tox-py37/96b83b2/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/49/636849/4/check/openstack-tox-docs/6b809d2/html/ : SUCCESS in 3m 25s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/4/check/tripleo-buildimage-overcloud-full-centos-7/232dd1e/ : SUCCESS in 25m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/4/check/tripleo-build-containers-centos-7/d7f52df/ : FAILURE in 17m 11s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/49/636849/4/check/tripleo-build-containers-centos-7-stein/17d8ed1/ : FAILURE in 16m 25s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/49/636849/4/check/tripleo-build-containers-centos-7-rocky/f30eab6/ : SUCCESS in 44m 26s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/4/check/tripleo-build-containers-centos-7-buildah/739e305/ : FAILURE in 17m 25s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/49/636849/4/check/tripleo-build-containers-centos-7-buildah-stein/900bf9c/ : FAILURE in 16m 32s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/4/check/tripleo-ci-centos-7-standalone/0007a63/ : SUCCESS in 1h 40m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/4/check/tripleo-ci-fedora-28-standalone/5cbd588/ : SUCCESS in 1h 47m 18s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/4/check/tripleo-ci-centos-7-standalone-upgrade/b8cd497/ : SUCCESS in 1h 49m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/10d9bbd/ : SUCCESS in 1h 58m 14s","accounts_in_message":[],"_revision_number":4},{"id":"e799706dba29b2613dcfcb50df6da0e2bb50f879","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-20 13:30:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"578fa1e0b1a39565be1d8f6cfe37c0f7910a7026","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 16:12:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/636849/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c3d0549/ : SUCCESS in 2h 28m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/636849/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a16b5aa/ : SUCCESS in 2h 36m 00s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/05d733a/ : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":5},{"id":"33c3b40bd1aec886df5a957038eeea6715701676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:59:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/49/636849/5/check/requirements-check/d619eb2/ : FAILURE in 2m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/5/check/openstack-tox-lower-constraints/9b00c6a/ : SUCCESS in 5m 09s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/5/check/openstackclient-check-plugins/f5b900f/ : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/5/check/openstack-tox-pep8/25de6be/ : SUCCESS in 9m 41s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/5/check/openstack-tox-py27/4a10e37/ : SUCCESS in 9m 34s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/5/check/openstack-tox-py35/a709a82/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/5/check/openstack-tox-py36/45de932/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/5/check/openstack-tox-py37/24e46df/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/49/636849/5/check/openstack-tox-docs/3f95f2c/html/ : SUCCESS in 3m 52s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/5/check/tripleo-buildimage-overcloud-full-centos-7/3d3667d/ : SUCCESS in 22m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/5/check/tripleo-build-containers-centos-7/a5893f4/ : SUCCESS in 51m 33s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/49/636849/5/check/tripleo-build-containers-centos-7-stein/a1dd8d1/ : SUCCESS in 51m 27s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/49/636849/5/check/tripleo-build-containers-centos-7-rocky/d08330d/ : SUCCESS in 45m 40s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/5/check/tripleo-build-containers-centos-7-buildah/628bc0d/ : SUCCESS in 57m 44s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/49/636849/5/check/tripleo-build-containers-centos-7-buildah-stein/2782d19/ : SUCCESS in 1h 01m 41s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/5/check/tripleo-ci-centos-7-standalone/0c006ae/ : SUCCESS in 1h 58m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/5/check/tripleo-ci-fedora-28-standalone/b9e93af/ : SUCCESS in 2h 09m 43s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/5/check/tripleo-ci-centos-7-standalone-upgrade/432e877/ : SUCCESS in 2h 17m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/1270219/ : SUCCESS in 2h 07m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8c2e604cd583031c9288c65fea809b5f4ba33fc7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-20 19:23:23.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e4fbc91f4e5bf08dffa6802b896e1b42e8ea621b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 22:00:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/636849/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e48839c/ : SUCCESS in 2h 36m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/636849/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/33e649f/ : SUCCESS in 2h 35m 41s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/49/636849/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/70671a5/ : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":6},{"id":"abc9a2b837ae889ad61d20156545c13d90ee8d0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:02:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/49/636849/6/check/requirements-check/d7b3b11/ : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/6/check/openstack-tox-lower-constraints/7c0003c/ : SUCCESS in 5m 32s\n- openstackclient-check-plugins http://logs.openstack.org/49/636849/6/check/openstackclient-check-plugins/58bc51b/ : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/6/check/openstack-tox-pep8/9e5a0ee/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/6/check/openstack-tox-py27/43d935f/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/6/check/openstack-tox-py35/faa5b2f/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/6/check/openstack-tox-py36/294fe4e/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/6/check/openstack-tox-py37/a913556/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/49/636849/6/check/openstack-tox-docs/e06f284/html/ : SUCCESS in 3m 15s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/6/check/tripleo-buildimage-overcloud-full-centos-7/033a30f/ : SUCCESS in 24m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/6/check/tripleo-build-containers-centos-7/bec42ef/ : SUCCESS in 45m 16s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/49/636849/6/check/tripleo-build-containers-centos-7-stein/5191ef0/ : SUCCESS in 53m 49s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/49/636849/6/check/tripleo-build-containers-centos-7-rocky/1f21b8a/ : SUCCESS in 41m 44s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/6/check/tripleo-build-containers-centos-7-buildah/667b23d/ : SUCCESS in 1h 02m 10s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/49/636849/6/check/tripleo-build-containers-centos-7-buildah-stein/0878c1f/ : SUCCESS in 58m 55s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/6/check/tripleo-ci-centos-7-standalone/5957a2a/ : SUCCESS in 1h 55m 35s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/6/check/tripleo-ci-fedora-28-standalone/cb76558/ : SUCCESS in 1h 54m 41s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/636849/6/check/tripleo-ci-centos-7-standalone-upgrade/b61a5d0/ : SUCCESS in 2h 00m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/3b321c7/ : SUCCESS in 2h 08m 03s","accounts_in_message":[],"_revision_number":6},{"id":"c0ebd1c69276d1f5a00942cbb4aabe6d8ee8ff17","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-21 02:44:20.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"fdb9643c5acb539e0200d7088110c33420c08f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:48:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"10d505911ebae2eb8b1b31807555c705b205b67f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:51:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/49/636849/6/gate/requirements-check/c350804/ : SUCCESS in 2m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/636849/6/gate/openstack-tox-lower-constraints/d3d2dd5/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/49/636849/6/gate/openstack-tox-pep8/4a53882/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/49/636849/6/gate/openstack-tox-py27/2d8e6df/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/49/636849/6/gate/openstack-tox-py35/3deaf49/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/49/636849/6/gate/openstack-tox-py36/40ae2c6/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/49/636849/6/gate/openstack-tox-py37/df25782/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/49/636849/6/gate/openstack-tox-docs/e0269f8/html/ : SUCCESS in 3m 10s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/636849/6/gate/tripleo-buildimage-overcloud-full-centos-7/6b9d862/ : SUCCESS in 23m 09s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/636849/6/gate/tripleo-build-containers-centos-7/0d7cbf3/ : SUCCESS in 44m 30s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/636849/6/gate/tripleo-build-containers-centos-7-buildah/49904e8/ : SUCCESS in 59m 39s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/636849/6/gate/tripleo-ci-centos-7-standalone/e9d8ddf/ : SUCCESS in 1h 54m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/636849/6/gate/tripleo-ci-fedora-28-standalone/7e0dd4b/ : SUCCESS in 1h 54m 29s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/49/636849/6/gate/tripleo-ci-centos-7-standalone-upgrade-stein/f141d40/ : SUCCESS in 2h 09m 16s","accounts_in_message":[],"_revision_number":6},{"id":"8f7b71f336f2b2ca0f2149a0a02496ed30ae92ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:51:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c880c8fde8dd24e9fc3c1821bfa2e0daa9bd0b7c","revisions":{"dc3ef46299cd57d406367a35f1b39ab1af440a86":{"kind":"REWORK","_number":1,"created":"2019-02-14 07:56:46.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/49/636849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/1"}}},"commit":{"parents":[{"commit":"623f841ab997a2c0587b3447f7ea78c61890e57f","subject":"Merge \"get_parser() should return a parser\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/623f841ab997a2c0587b3447f7ea78c61890e57f"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/dc3ef46299cd57d406367a35f1b39ab1af440a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/dc3ef46299cd57d406367a35f1b39ab1af440a86"}]},"branch":"refs/heads/master"},"006bfbadceca8045ab4d5a9671aa74bfdd70495f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-07 19:20:16.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/636849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/2"}}},"commit":{"parents":[{"commit":"91f621081083d1433c9b93fb8d5224ad010860bd","subject":"Merge \"Implement tripleo container image list command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/91f621081083d1433c9b93fb8d5224ad010860bd"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-07 19:20:16.000000000","tz":0},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/006bfbadceca8045ab4d5a9671aa74bfdd70495f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/006bfbadceca8045ab4d5a9671aa74bfdd70495f"}]},"branch":"refs/heads/master"},"8227fa6976777c4aaf8d366e5fd07ef9062d2335":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-19 21:01:37.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/636849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/3"}}},"commit":{"parents":[{"commit":"3689b4075d10cde753e20109d15c25fdc64f961b","subject":"Fixup sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3689b4075d10cde753e20109d15c25fdc64f961b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-19 21:01:37.000000000","tz":0},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8227fa6976777c4aaf8d366e5fd07ef9062d2335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8227fa6976777c4aaf8d366e5fd07ef9062d2335"}]},"branch":"refs/heads/master"},"1a379e1333f2fa504a08c03a71c76f6cd273c182":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-19 21:05:55.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/636849/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/4"}}},"commit":{"parents":[{"commit":"3435ec5f5567bd07f9bedda06a18b1f7c932b937","subject":"Fixup sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3435ec5f5567bd07f9bedda06a18b1f7c932b937"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-19 21:05:55.000000000","tz":0},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a379e1333f2fa504a08c03a71c76f6cd273c182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a379e1333f2fa504a08c03a71c76f6cd273c182"}]},"branch":"refs/heads/master"},"1be32fc67db865afba6687828fc6585fbea2454c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-20 13:30:05.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/636849/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/5"}}},"commit":{"parents":[{"commit":"fa98acc72b91be559646cd0ccca656f69543d31e","subject":"Fixup sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fa98acc72b91be559646cd0ccca656f69543d31e"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-20 13:30:05.000000000","tz":0},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1be32fc67db865afba6687828fc6585fbea2454c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1be32fc67db865afba6687828fc6585fbea2454c"}]},"branch":"refs/heads/master"},"c880c8fde8dd24e9fc3c1821bfa2e0daa9bd0b7c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-20 19:23:23.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/49/636849/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/636849/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/636849/6"}}},"commit":{"parents":[{"commit":"81d26455ab71f1aa3d6956e6a0eaadd822d976df","subject":"Fixup sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/81d26455ab71f1aa3d6956e6a0eaadd822d976df"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-02-14 07:54:20.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-20 19:23:23.000000000","tz":0},"subject":"Bump keystone middleware version to 4.18","message":"Bump keystone middleware version to 4.18\n\nkeystone_authtoken/auth_uri is deprecated. Use www_authenticate_uri\ninstead.\n\nThis is related to this bug https://bugs.launchpad.net/mistral/+bug/1788174\n\nkeystonemiddleware in lower constraints should be increased\nbecause www_authenticate_uri was introduced in keystonemiddleware 4.18.0.\n\nChange-Id: I427887467678c11c74382b6805e32d1455641294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c880c8fde8dd24e9fc3c1821bfa2e0daa9bd0b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c880c8fde8dd24e9fc3c1821bfa2e0daa9bd0b7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
