)]}'
{"id":"openstack%2Fpython-tripleoclient~651567","triplet_id":"openstack%2Fpython-tripleoclient~master~Ib1dc0c08ce7971a03639acc42b1e738d93a52f98","project":"openstack/python-tripleoclient","branch":"master","topic":"convert_docker_params","hashtags":[],"change_id":"Ib1dc0c08ce7971a03639acc42b1e738d93a52f98","subject":"Wire in convert_docker_params","status":"MERGED","created":"2019-04-10 14:48:36.000000000","updated":"2019-06-16 03:30:42.000000000","submitted":"2019-06-16 03:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651567-1560655842974-018dc456","meta_rev_id":"af63ec35d7c70404a3ad63a01fa16ba73920c39d","_number":651567,"virtual_id_number":651567,"owner":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-06-16 03:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-12 18:38:20.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-12 15:14:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":2,"date":"2019-06-16 01:10:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":1,"date":"2019-06-16 01:10:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 15:04:40.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-06-04 18:38:18.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-06-06 13:43:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-12 18:38:20.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-16 01:10:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-16 03:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e11ed292cb5bdeddc2424fe94e7c882541c67f6","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-04-10 14:48:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e782127d5503fd5e93eb730cd7b6a85914340fe3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-10 15:18:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/09f6981/ : FAILURE in 25m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/359b8ae/ : FAILURE in 25m 07s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/1224e03/ : FAILURE in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ee0b27ebe8da10b2b6888de8d619d9cdbc6643cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 15:26:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/1/check/openstack-tox-lower-constraints/e18a518/ : FAILURE in 5m 52s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/1/check/openstackclient-check-plugins/438ef60/ : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/1/check/openstack-tox-pep8/d3fc766/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/1/check/openstack-tox-py27/eb8dffb/ : FAILURE in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/1/check/openstack-tox-py35/f7d3b52/ : FAILURE in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/1/check/openstack-tox-py36/c0d277f/ : FAILURE in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/1/check/openstack-tox-py37/058d05b/ : FAILURE in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/67/651567/1/check/openstack-tox-docs/b74b009/ : FAILURE in 5m 30s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/1/check/tripleo-buildimage-overcloud-full-centos-7/05a5c1f/ : SUCCESS in 25m 11s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/1/check/tripleo-ci-centos-7-undercloud-containers/54f223d/ : FAILURE in 29m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/1/check/tripleo-ci-centos-7-standalone/f5604e2/ : FAILURE in 24m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/1/check/tripleo-ci-fedora-28-standalone/81c963a/ : FAILURE in 29m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b582448990b3f8b493fde4d0e23af37175897334","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-04-10 17:26:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93185bb1d284eaaae754836d62d0f6e127ce2d16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-10 17:54:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bb99c6d/ : FAILURE in 24m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/997e859/ : FAILURE in 22m 15s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 finger://ze04.softwarefactory-project.io:7979/fc3b46cbc92e425eb5c019658da28c27 : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":1},{"id":"28471c7cfd3d9f558ec588b6b1c317f900210bd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 17:59:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/1/check/openstack-tox-lower-constraints/9cd0174/ : FAILURE in 6m 19s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/1/check/openstackclient-check-plugins/3478630/ : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/1/check/openstack-tox-pep8/2442d5c/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/1/check/openstack-tox-py27/6213a8f/ : FAILURE in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/1/check/openstack-tox-py35/ac6ee3a/ : FAILURE in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/1/check/openstack-tox-py36/b28baf9/ : FAILURE in 2m 55s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/1/check/openstack-tox-py37/b919e58/ : FAILURE in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/67/651567/1/check/openstack-tox-docs/77a971c/ : FAILURE in 5m 56s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/1/check/tripleo-buildimage-overcloud-full-centos-7/d56eb00/ : SUCCESS in 23m 53s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/1/check/tripleo-ci-centos-7-undercloud-containers/82e667a/ : FAILURE in 24m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/1/check/tripleo-ci-centos-7-standalone/cb183a0/ : FAILURE in 19m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/1/check/tripleo-ci-fedora-28-standalone/0b4de30/ : FAILURE in 26m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7af760cd968e988afe416ed0d42cbbf99456612f","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-04-10 20:30:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"38a9ade4abc319749182118e118111c0e5a0a5d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-10 20:55:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/885bc1c/ : FAILURE in 21m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f793b66/ : FAILURE in 20m 34s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5a26338/ : FAILURE in 17m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3bdf3d7c9bac632116a962683184f33100a58068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 21:06:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/2/check/openstack-tox-lower-constraints/a923a16/ : FAILURE in 5m 03s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/2/check/openstackclient-check-plugins/834eb2a/ : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/2/check/openstack-tox-pep8/6be92b4/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/2/check/openstack-tox-py27/0b2efde/ : FAILURE in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/2/check/openstack-tox-py35/b82d174/ : FAILURE in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/2/check/openstack-tox-py36/0b2f1fe/ : FAILURE in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/2/check/openstack-tox-py37/5d508ab/ : FAILURE in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/67/651567/2/check/openstack-tox-docs/7570330/ : FAILURE in 3m 51s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/2/check/tripleo-buildimage-overcloud-full-centos-7/2085d24/ : SUCCESS in 22m 26s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/2/check/tripleo-ci-centos-7-undercloud-containers/258dc16/ : FAILURE in 25m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/2/check/tripleo-ci-centos-7-standalone/ed8f653/ : FAILURE in 26m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/2/check/tripleo-ci-fedora-28-standalone/cdfdbbd/ : FAILURE in 29m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5a4eb5717554c03a30ea25dd3c4b44c10cb3bc4","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-04-10 23:40:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed7a77fdbac76a2b37414f98de815a784a3e29b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 02:02:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/3/check/openstack-tox-lower-constraints/bb8e3c0/ : FAILURE in 4m 09s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/3/check/openstackclient-check-plugins/6236f6f/ : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/3/check/openstack-tox-pep8/c9657ab/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/3/check/openstack-tox-py27/25d19bb/ : FAILURE in 2m 52s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/3/check/openstack-tox-py35/f7281c6/ : FAILURE in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/3/check/openstack-tox-py36/d7b4aea/ : FAILURE in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/3/check/openstack-tox-py37/13915fd/ : FAILURE in 3m 48s\n- openstack-tox-docs http://logs.openstack.org/67/651567/3/check/openstack-tox-docs/3e8b741/html/ : SUCCESS in 3m 09s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/3/check/tripleo-buildimage-overcloud-full-centos-7/59c0454/ : SUCCESS in 17m 43s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/3/check/tripleo-ci-centos-7-undercloud-containers/f2ae918/ : SUCCESS in 1h 30m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/3/check/tripleo-ci-centos-7-standalone/a2f9658/ : SUCCESS in 1h 36m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/3/check/tripleo-ci-fedora-28-standalone/903578c/ : SUCCESS in 2h 15m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf804170d1f6aa26022ea48a89d8ccae7ae66dd4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-11 02:44:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/de2cba8/ : SUCCESS in 2h 48m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1a220f0/ : SUCCESS in 3h 00m 35s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/51f2ee9/ : SUCCESS in 1h 28m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d52f4af942e72a45a927d8037eb9e032b8a0e177","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-04-16 16:04:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"38fd7daf4edf6a5f8f26e67ac4395f7a650da150","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-16 19:06:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/cfa2d51323dc4d6f8095032c18bab5c2 : RETRY_LIMIT in 8s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1ca828c/ : SUCCESS in 2h 55m 30s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/744ee78/ : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":3},{"id":"34693cac922bc0dddee4ec52760c58917cc046d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 19:34:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/3/check/openstack-tox-lower-constraints/64fdb9b/ : FAILURE in 5m 23s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/3/check/openstackclient-check-plugins/d110e6a/ : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/3/check/openstack-tox-pep8/0e12161/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/3/check/openstack-tox-py27/db39934/ : FAILURE in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/3/check/openstack-tox-py35/353c783/ : FAILURE in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/3/check/openstack-tox-py36/00cf8b2/ : FAILURE in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/3/check/openstack-tox-py37/5c33cdd/ : FAILURE in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/67/651567/3/check/openstack-tox-docs/e4dc119/html/ : SUCCESS in 3m 43s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/3/check/tripleo-buildimage-overcloud-full-centos-7/0c0225a/ : SUCCESS in 22m 38s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/3/check/tripleo-ci-centos-7-undercloud-containers/45c8151/ : SUCCESS in 1h 25m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/3/check/tripleo-ci-centos-7-standalone/1e4df68/ : SUCCESS in 1h 50m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/3/check/tripleo-ci-fedora-28-standalone/04c80be/ : FAILURE in 2h 00m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"801e34761f5b86c1c2ea707a0bdb9f92d7efc8cf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-23 13:21:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ad953bcce340d12373f62bd14ede1aa44f4d8105","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-23 13:22:25.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9e813903fc4c22be47e9880847fcdf160d331f9a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-23 13:22:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"57a5b401a00ee2d90643126c46837ba9aaeab0a9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-23 16:45:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/265e48a/ : FAILURE in 3h 20m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/843b46b/ : FAILURE in 3h 04m 35s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b21a8b1/ : SUCCESS in 1h 42m 35s","accounts_in_message":[],"_revision_number":5},{"id":"909f652cbf439c7f3f75de4ddd4b94b8b5ed8a5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 01:45:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/5/check/openstack-tox-lower-constraints/dc25c48/ : FAILURE in 4m 35s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/5/check/openstackclient-check-plugins/3c93528/ : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/5/check/openstack-tox-pep8/79d4fd7/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/5/check/openstack-tox-py27/be90c72/ : FAILURE in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/5/check/openstack-tox-py35/43097f9/ : FAILURE in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/5/check/openstack-tox-py36/63a8100/ : FAILURE in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/5/check/openstack-tox-py37/21dd078/ : FAILURE in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/67/651567/5/check/openstack-tox-docs/6828802/ : POST_FAILURE in 3m 41s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/5/check/tripleo-buildimage-overcloud-full-centos-7/c75a6d5/ : SUCCESS in 22m 05s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/5/check/tripleo-ci-centos-7-undercloud-containers/41636fe/ : SUCCESS in 1h 23m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/5/check/tripleo-ci-centos-7-standalone/f27a4a8/ : SUCCESS in 1h 46m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/5/check/tripleo-ci-fedora-28-standalone/6c0d6dd/ : SUCCESS in 1h 40m 35s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/67/651567/5/check/tripleo-ci-centos-7-standalone-rocky/6d2755a/ : SUCCESS in 1h 34m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"138c4391d5351ba7e07ccc1cb37e51133f31ae88","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-24 21:47:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4f710fea42b48fcb8edab8d257e7dabfb20a73da","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-04-25 07:34:07.000000000","message":"Patch Set 5: Code-Review-1\n\nYou need to bump up tripleo-common version in requirements to 10.7","accounts_in_message":[],"_revision_number":5},{"id":"75981aeb656b81c7cad30e939b64fbef95026a11","author":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"date":"2019-05-13 13:29:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0b167e5ca6428c09e836372e73844b9975daa865","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-13 15:04:40.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b05e8f11cf2cfb6d655353041b6435d617d6e2c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 16:34:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c3578b9/ : SUCCESS in 3h 03m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6ac1a76/ : SUCCESS in 3h 03m 34s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5ccabbc/ : SUCCESS in 1h 36m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a82d6c838634de0b7b73e1cdc0befa1ec63a8a9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 16:51:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/651567/6/check/requirements-check/821d954/ : FAILURE in 1m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/6/check/openstack-tox-lower-constraints/83ba8a7/ : SUCCESS in 4m 09s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/6/check/openstackclient-check-plugins/acbfb53/ : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/6/check/openstack-tox-pep8/b9b44b8/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/6/check/openstack-tox-py27/7a75939/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/6/check/openstack-tox-py35/a896bb0/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/6/check/openstack-tox-py36/d26187d/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/6/check/openstack-tox-py37/4195253/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/67/651567/6/check/openstack-tox-docs/c5f9f37/html/ : SUCCESS in 3m 05s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/6/check/tripleo-buildimage-overcloud-full-centos-7/8057b32/ : SUCCESS in 21m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/6/check/tripleo-build-containers-centos-7/91521d7/ : SUCCESS in 42m 21s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/6/check/tripleo-build-containers-centos-7-stein/ed1d814/ : SUCCESS in 45m 49s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/6/check/tripleo-build-containers-centos-7-rocky/91edbeb/ : SUCCESS in 35m 53s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/6/check/tripleo-build-containers-centos-7-buildah/a455713/ : SUCCESS in 51m 09s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/6/check/tripleo-build-containers-centos-7-buildah-stein/6c96d45/ : SUCCESS in 57m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-containers-multinode/5e7bd5f/ : SUCCESS in 1h 56m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-undercloud-containers/48b3fd7/ : SUCCESS in 1h 30m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2c3e6a3/ : SUCCESS in 2h 14m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/713b51b/ : FAILURE in 2h 57m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-standalone/a929328/ : SUCCESS in 1h 41m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/6/check/tripleo-ci-fedora-28-standalone/c870f4b/ : SUCCESS in 1h 41m 46s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-standalone-upgrade/bc2fe21/ : SUCCESS in 2h 17m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/1b0dfa5/ : SUCCESS in 1h 55m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/98a3e02/ : FAILURE in 1h 47m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-standalone-rocky/db60a4c/ : SUCCESS in 1h 30m 20s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/7ddd59c/ : SUCCESS in 2h 11m 40s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/2279051/ : FAILURE in 1h 15m 26s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/c7a3982/ : SUCCESS in 1h 58m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f966edb3d6b733536fd67c1902d3934b05fa6a11","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-05-16 08:36:49.000000000","message":"Patch Set 6: Code-Review-1\n\nyou need also update lower-constraints.txt with tripleo-common\u003d\u003d10.7","accounts_in_message":[],"_revision_number":6},{"id":"b937b5f7631cdeb41ce95387cc9113eff353c7d5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 17:51:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6d9512d59f577be946caad8e24eea58f52069bfa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 20:20:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3d5233f/ : SUCCESS in 2h 24m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8551fb8/ : SUCCESS in 2h 25m 34s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/9ed48ff/ : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":7},{"id":"e3e84bd58446224f941badb9c065743f9f7555f6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-31 20:25:26.000000000","message":"Patch Set 7: Code-Review+1\n\nwe need first https://review.rdoproject.org/r/20986 or we\u0027ll break things","accounts_in_message":[],"_revision_number":7},{"id":"cbb581dce254d4d08dcfaba05b2a8f6f491d0ecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 23:20:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/651567/7/check/requirements-check/e45deb0/ : FAILURE in 3m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/7/check/openstack-tox-lower-constraints/6593a5c/ : SUCCESS in 4m 45s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/7/check/openstackclient-check-plugins/64def30/ : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/7/check/openstack-tox-pep8/829f0d6/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/7/check/openstack-tox-py27/35a71be/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/7/check/openstack-tox-py35/2b3830a/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/7/check/openstack-tox-py36/960c06d/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/7/check/openstack-tox-py37/0047d94/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/67/651567/7/check/openstack-tox-docs/db36d94/html/ : SUCCESS in 2m 59s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/7/check/tripleo-buildimage-overcloud-full-centos-7/65a03e0/ : SUCCESS in 25m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/7/check/tripleo-build-containers-centos-7/b2f404c/ : SUCCESS in 40m 36s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/7/check/tripleo-build-containers-centos-7-stein/5db22bf/ : SUCCESS in 56m 34s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/7/check/tripleo-build-containers-centos-7-rocky/892c5fa/ : SUCCESS in 35m 30s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/7/check/tripleo-build-containers-centos-7-buildah/a46e832/ : SUCCESS in 58m 31s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/7/check/tripleo-build-containers-centos-7-buildah-stein/a6d44a6/ : SUCCESS in 43m 03s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-containers-multinode/54d23bb/ : SUCCESS in 2h 08m 00s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-undercloud-containers/23c4d2e/ : SUCCESS in 1h 33m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4299e70/ : SUCCESS in 2h 17m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/712c0fd/ : SUCCESS in 2h 13m 23s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-standalone/d156283/ : SUCCESS in 1h 50m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/7/check/tripleo-ci-fedora-28-standalone/950c43f/ : SUCCESS in 1h 59m 51s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-standalone-upgrade/a7dc51b/ : SUCCESS in 2h 08m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/30f62e5/ : SUCCESS in 1h 52m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1567edf/ : FAILURE in 1h 48m 59s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a71fa89/ : SUCCESS in 2h 08m 40s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/8de096e/ : FAILURE in 1h 33m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4344b75/ : SUCCESS in 2h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5b070921e3d73ce572769a54fcbd68039eba8e56","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-03 14:41:27.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"9ec9cc6a99394158d8c8399c917c6f4e199cd770","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 17:39:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/70b4eca/ : SUCCESS in 2h 56m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d67217a/ : SUCCESS in 2h 52m 07s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/8/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/df1015c/ : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":8},{"id":"d84985fc5c78936b837f7fb25bb32dbd26463d99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:52:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/651567/8/check/requirements-check/82d351d/ : FAILURE in 2m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/8/check/openstack-tox-lower-constraints/6766902/ : SUCCESS in 4m 44s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/8/check/openstackclient-check-plugins/10497b5/ : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/8/check/openstack-tox-pep8/b984f2f/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/8/check/openstack-tox-py27/b7ec484/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/8/check/openstack-tox-py35/80f7771/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/8/check/openstack-tox-py36/41eeb1f/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/8/check/openstack-tox-py37/f61ad9e/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/67/651567/8/check/openstack-tox-docs/626a953/html/ : SUCCESS in 4m 15s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/8/check/tripleo-buildimage-overcloud-full-centos-7/c10876d/ : SUCCESS in 22m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/8/check/tripleo-build-containers-centos-7/d0da09b/ : SUCCESS in 51m 13s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/8/check/tripleo-build-containers-centos-7-stein/0111ec3/ : SUCCESS in 37m 07s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/8/check/tripleo-build-containers-centos-7-rocky/be4d0c3/ : SUCCESS in 43m 12s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/8/check/tripleo-build-containers-centos-7-buildah/94fadc4/ : SUCCESS in 1h 10m 03s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/8/check/tripleo-build-containers-centos-7-buildah-stein/6721b1d/ : SUCCESS in 48m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-containers-multinode/6b19aff/ : SUCCESS in 2h 03m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-undercloud-containers/edb866a/ : SUCCESS in 1h 45m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9263c17/ : SUCCESS in 2h 20m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ff338c4/ : SUCCESS in 2h 33m 49s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-standalone/1ada01c/ : SUCCESS in 1h 54m 24s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/8/check/tripleo-ci-fedora-28-standalone/e3c645a/ : SUCCESS in 1h 49m 41s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-standalone-upgrade/cd63b03/ : SUCCESS in 2h 24m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-standalone-upgrade-stein/4d0d415/ : SUCCESS in 2h 13m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/352071c/ : FAILURE in 1h 21m 47s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a2042e1/ : SUCCESS in 2h 12m 59s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/33fbcf8/ : FAILURE in 1h 18m 23s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/8/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/e447f17/ : SUCCESS in 2h 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"57b5f9c6e101266a75d3c19f304ddc979fe1acf5","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-06-04 18:38:18.000000000","message":"Patch Set 8:\n\nThis patch will not pass zuul until this https://review.opendev.org/#/c/663117/ is merged","accounts_in_message":[],"_revision_number":8},{"id":"2836c64cc35a871def5eea1596fc2e681fbf750c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 17:09:07.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a78ddb7da178afadd8aeabe9ea354ff79fc26933","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 19:37:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8350f05/ : SUCCESS in 2h 18m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/688a4fa/ : SUCCESS in 2h 27m 14s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/9/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/212abf5/ : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":9},{"id":"6850c1405ebe8b6c5dbd4fc72e8edca8863e4159","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 20:42:07.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"41e373ae9e347f90adb5abe776aead0c7e24722b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 23:09:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7b4abfd/ : SUCCESS in 2h 19m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/73877d5/ : SUCCESS in 2h 25m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/10/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c099dcb/ : SUCCESS in 1h 13m 03s","accounts_in_message":[],"_revision_number":10},{"id":"3cbf53d117e698cc8bb93f21b28c0daadd16c9d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:39:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/651567/10/check/requirements-check/d2e4b8d/ : FAILURE in 2m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/10/check/openstack-tox-lower-constraints/22a836a/ : SUCCESS in 6m 22s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/10/check/openstackclient-check-plugins/39917b4/ : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/10/check/openstack-tox-pep8/91641e7/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/10/check/openstack-tox-py27/a706b9e/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/10/check/openstack-tox-py35/7309fea/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/10/check/openstack-tox-py36/ac4f856/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/10/check/openstack-tox-py37/8469fb0/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/67/651567/10/check/openstack-tox-docs/4da565d/html/ : SUCCESS in 3m 56s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/10/check/tripleo-buildimage-overcloud-full-centos-7/5f3fa80/ : SUCCESS in 17m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/10/check/tripleo-build-containers-centos-7/153f8d4/ : SUCCESS in 35m 29s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/10/check/tripleo-build-containers-centos-7-stein/e34555b/ : SUCCESS in 55m 55s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/10/check/tripleo-build-containers-centos-7-rocky/908964f/ : SUCCESS in 34m 11s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/10/check/tripleo-build-containers-centos-7-buildah/6326239/ : SUCCESS in 57m 51s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/10/check/tripleo-build-containers-centos-7-buildah-stein/f32a54a/ : SUCCESS in 50m 43s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-containers-multinode/65c0a48/ : SUCCESS in 2h 16m 44s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-undercloud-containers/3ad7eff/ : SUCCESS in 1h 28m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d33ff62/ : SUCCESS in 2h 13m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b931daa/ : SUCCESS in 2h 34m 02s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-standalone/a592187/ : SUCCESS in 2h 02m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/10/check/tripleo-ci-fedora-28-standalone/68f26ae/ : SUCCESS in 1h 44m 05s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-standalone-upgrade/ed8fce6/ : SUCCESS in 1h 59m 30s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-standalone-upgrade-stein/7bffa69/ : SUCCESS in 2h 23m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3b440e1/ : FAILURE in 1h 33m 22s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/1bfd648/ : SUCCESS in 2h 09m 18s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/eb69008/ : FAILURE in 1h 21m 51s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/10/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/ca3805c/ : SUCCESS in 2h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4118b7054deb4ba7fc060691758466c05837b8e4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:24:52.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"15cc90cefddf85fffb4b89ed2d4e9cf4399a0487","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:25:29.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"38d7b2962f2911c04df010188326e81264d0a9e9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:25:42.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"48f42ec2a172c8da26b8f3a6cb46c4fe34895bcd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:43:11.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"146e362b9057bade9fd81ce95e75baf6704d47a0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 15:57:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ebc823a/ : SUCCESS in 2h 23m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/341831e/ : SUCCESS in 2h 30m 29s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/12/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/f7cca4b/ : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":12},{"id":"00f1422343c8c05fb7aaf079eb8ed4677ebe7d49","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 18:51:41.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"caa7d9693ced8e7809f84f19b9a5ad2c4d7001f1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 21:24:02.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d497fdb/ : SUCCESS in 2h 20m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ea45117/ : SUCCESS in 2h 30m 26s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/13/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/57a4342/ : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":13},{"id":"89bd9936f2e59cbd8b9c001e7a90dcc1b911e898","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:42:44.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/651567/13/check/requirements-check/95bbcb8/ : FAILURE in 2m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/13/check/openstack-tox-lower-constraints/da60fe1/ : SUCCESS in 5m 29s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/13/check/openstackclient-check-plugins/09b6d6b/ : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/13/check/openstack-tox-pep8/7ab11f0/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/13/check/openstack-tox-py27/7f7fe1a/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/13/check/openstack-tox-py35/0c9ea12/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/13/check/openstack-tox-py36/3cf6791/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/13/check/openstack-tox-py37/6eea0c0/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/67/651567/13/check/openstack-tox-docs/cec1d91/html/ : SUCCESS in 3m 25s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/13/check/tripleo-buildimage-overcloud-full-centos-7/b4442f3/ : SUCCESS in 18m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/13/check/tripleo-build-containers-centos-7/d1fdb49/ : SUCCESS in 38m 06s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/13/check/tripleo-build-containers-centos-7-stein/5bdf596/ : SUCCESS in 48m 03s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/13/check/tripleo-build-containers-centos-7-rocky/2010549/ : SUCCESS in 46m 53s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/13/check/tripleo-build-containers-centos-7-buildah/742228c/ : SUCCESS in 51m 20s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/13/check/tripleo-build-containers-centos-7-buildah-stein/84da81e/ : SUCCESS in 50m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-containers-multinode/840cf11/ : SUCCESS in 2h 05m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-undercloud-containers/70d89d5/ : SUCCESS in 1h 27m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4d0623e/ : SUCCESS in 2h 15m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/24f1a31/ : SUCCESS in 2h 37m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-standalone/a29b9df/ : SUCCESS in 2h 02m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/13/check/tripleo-ci-fedora-28-standalone/588fd37/ : SUCCESS in 2h 00m 01s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-standalone-upgrade/f8cc217/ : SUCCESS in 2h 13m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-standalone-upgrade-stein/bc45453/ : SUCCESS in 2h 14m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/18edb4a/ : FAILURE in 1h 27m 13s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/902427c/ : SUCCESS in 2h 06m 43s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/1846742/ : FAILURE in 1h 18m 14s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/13/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6163ab4/ : SUCCESS in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1dd84c3f0e396aca1062cdba23d442a39696b0ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-11 13:23:49.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"6f6f5fd1c1f1eed59b3c7e53f8d69cf1b5f4f58e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 16:32:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/16b3234/ : SUCCESS in 3h 08m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0223787/ : SUCCESS in 2h 56m 45s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/14/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5d8306f/ : SUCCESS in 1h 25m 43s","accounts_in_message":[],"_revision_number":14},{"id":"e8b6a3521ce7c58171ae35d49f361de0b647e725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:23:22.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/67/651567/14/check/requirements-check/0873468/ : SUCCESS in 2m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/14/check/openstack-tox-lower-constraints/17ddbca/ : SUCCESS in 5m 33s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/14/check/openstackclient-check-plugins/8702219/ : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/14/check/openstack-tox-pep8/06679b2/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/14/check/openstack-tox-py27/f13ce20/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/14/check/openstack-tox-py35/00ef4f1/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/14/check/openstack-tox-py36/d3394ec/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/14/check/openstack-tox-py37/0d90ee6/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/67/651567/14/check/openstack-tox-docs/550c5f5/html/ : SUCCESS in 3m 36s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/14/check/tripleo-buildimage-overcloud-full-centos-7/2c208ae/ : SUCCESS in 24m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/14/check/tripleo-build-containers-centos-7/d13779e/ : SUCCESS in 44m 23s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/14/check/tripleo-build-containers-centos-7-stein/f42f696/ : SUCCESS in 45m 21s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/14/check/tripleo-build-containers-centos-7-rocky/ba6f24e/ : SUCCESS in 45m 30s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/14/check/tripleo-build-containers-centos-7-buildah/f2420e9/ : SUCCESS in 51m 13s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/14/check/tripleo-build-containers-centos-7-buildah-stein/da7deee/ : SUCCESS in 1h 03m 31s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-containers-multinode/a489d19/ : SUCCESS in 2h 07m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-undercloud-containers/732b300/ : SUCCESS in 2h 06m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2731cfa/ : SUCCESS in 2h 20m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ebea058/ : SUCCESS in 2h 29m 44s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-standalone/ee3093d/ : SUCCESS in 2h 28m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/14/check/tripleo-ci-fedora-28-standalone/c24ab3f/ : SUCCESS in 1h 49m 29s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-standalone-upgrade/46aefae/ : SUCCESS in 2h 00m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-standalone-upgrade-stein/a3f322d/ : SUCCESS in 1h 48m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/eb067d4/ : FAILURE in 1h 23m 57s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/894dd07/ : SUCCESS in 2h 18m 49s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/a565f20/ : FAILURE in 1h 24m 32s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/14/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/1d32ff4/ : SUCCESS in 2h 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ccb59b010697500f8a9eb90938ad6880e9d804cc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 01:03:47.000000000","message":"Patch Set 14: Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"dfb3a1d43f0ccd9f91ed557455b47624eb804a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:26:36.000000000","message":"Patch Set 14: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"cf43ea60e5fad5402c2f6ef88ac709b1eba79774","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-06-12 12:57:01.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"1f363141423fac770574af313ff5b78d925d4cac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 15:14:44.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"d9b21a5fec90b2e7f34e8a79fa67ea3cf2786e38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 18:38:20.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/67/651567/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f45b71f/ : SUCCESS in 2h 40m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/67/651567/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b189d0f/ : FAILURE in 1h 13m 56s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/67/651567/16/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/26b054d/ : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":16},{"id":"39095e4f9b15524de893d88129f9693b9cd7b0f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 05:55:01.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/67/651567/16/check/requirements-check/4ead738/ : SUCCESS in 1m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/16/check/openstack-tox-lower-constraints/5385d3f/ : SUCCESS in 4m 05s\n- openstackclient-check-plugins http://logs.openstack.org/67/651567/16/check/openstackclient-check-plugins/c89a735/ : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/16/check/openstack-tox-pep8/1c828e0/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/16/check/openstack-tox-py27/21eb617/ : SUCCESS in 2m 50s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/16/check/openstack-tox-py35/edaf39c/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/16/check/openstack-tox-py36/844440b/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/16/check/openstack-tox-py37/d448918/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/67/651567/16/check/openstack-tox-docs/93ecaf9/html/ : SUCCESS in 3m 29s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/16/check/tripleo-buildimage-overcloud-full-centos-7/f63ab13/ : SUCCESS in 23m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/16/check/tripleo-build-containers-centos-7/0a25d02/ : SUCCESS in 47m 04s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/16/check/tripleo-build-containers-centos-7-stein/7038118/ : SUCCESS in 52m 00s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/16/check/tripleo-build-containers-centos-7-rocky/565edc1/ : SUCCESS in 37m 55s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/16/check/tripleo-build-containers-centos-7-buildah/bdd4707/ : SUCCESS in 1h 00m 02s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/16/check/tripleo-build-containers-centos-7-buildah-stein/b715a35/ : SUCCESS in 1h 01m 57s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-containers-multinode/e542dd6/ : SUCCESS in 2h 10m 47s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-undercloud-containers/c0920de/ : SUCCESS in 1h 41m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cfd6288/ : SUCCESS in 2h 19m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d742a9c/ : SUCCESS in 2h 35m 25s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-standalone/5243e21/ : SUCCESS in 1h 55m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/16/check/tripleo-ci-fedora-28-standalone/641237e/ : SUCCESS in 2h 05m 04s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-standalone-upgrade/e37a643/ : SUCCESS in 2h 05m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-standalone-upgrade-stein/e7d7cbc/ : SUCCESS in 2h 03m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a8b5c15/ : FAILURE in 1h 42m 45s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/9a40ef6/ : SUCCESS in 2h 21m 17s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/17038a9/ : FAILURE in 1h 27m 28s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/67/651567/16/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/e70c1ac/ : SUCCESS in 2h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"835dd6f9d466985a6bb80127252a2b8600c8252c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-16 01:10:12.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"1d359a317678215199d1a726a60284f1d48e50d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 01:10:27.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"eb2fe6fc97b16f9329a2b3b7e10ff2de0cde2049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 03:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"65a4094923bf7701d5ef14b171af5d1f88d4e12f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 03:30:42.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/67/651567/16/gate/requirements-check/92da92c/ : SUCCESS in 1m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/651567/16/gate/openstack-tox-lower-constraints/e71b107/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/67/651567/16/gate/openstack-tox-pep8/031b16f/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/67/651567/16/gate/openstack-tox-py27/5679b02/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/67/651567/16/gate/openstack-tox-py35/fab7774/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/67/651567/16/gate/openstack-tox-py36/c824fcf/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/67/651567/16/gate/openstack-tox-py37/95cd2c6/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/67/651567/16/gate/openstack-tox-docs/d4777c7/html/ : SUCCESS in 2m 53s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/651567/16/gate/tripleo-buildimage-overcloud-full-centos-7/2f06f0c/ : SUCCESS in 20m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/67/651567/16/gate/tripleo-build-containers-centos-7/c255b12/ : SUCCESS in 51m 27s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/67/651567/16/gate/tripleo-build-containers-centos-7-stein/de9e6e1/ : SUCCESS in 40m 48s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/67/651567/16/gate/tripleo-build-containers-centos-7-rocky/887d78f/ : SUCCESS in 30m 09s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/67/651567/16/gate/tripleo-build-containers-centos-7-buildah/75073dd/ : SUCCESS in 56m 37s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/67/651567/16/gate/tripleo-build-containers-centos-7-buildah-stein/440ce18/ : SUCCESS in 57m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-containers-multinode/2095a5d/ : SUCCESS in 2h 17m 29s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-standalone/b68797a/ : SUCCESS in 1h 45m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-fedora-28-standalone/f4434da/ : SUCCESS in 1h 46m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5b37b51/ : SUCCESS in 2h 09m 50s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-standalone-upgrade-stein/35f26e0/ : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-undercloud-containers/eff2e99/ : SUCCESS in 1h 26m 21s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-scenario001-standalone/f6a4084/ : SUCCESS in 1h 35m 50s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-scenario002-standalone/64d4fb0/ : SUCCESS in 1h 33m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-scenario003-standalone/dfb9438/ : SUCCESS in 1h 10m 30s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/67/651567/16/gate/tripleo-ci-centos-7-scenario004-standalone/519746e/ : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"50744f7c229096db8cf07199d16485d6a11338f6","revisions":{"5b21e726db41babd0d6130ba9a67b73c9a559219":{"kind":"REWORK","_number":1,"created":"2019-04-10 14:48:36.000000000","uploader":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"ref":"refs/changes/67/651567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/1"}}},"commit":{"parents":[{"commit":"d153846e0c2058e21b9161820e3c47b22f381141","subject":"Merge \"Add checking mechanism driver during upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d153846e0c2058e21b9161820e3c47b22f381141"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients. In the future\nwe can use it for a few other ad-hoc\nDocker* variable substitutions as well.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\nDepends-On: I29d5791acaafa31de15bd58a79302a6b0cb0b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5b21e726db41babd0d6130ba9a67b73c9a559219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5b21e726db41babd0d6130ba9a67b73c9a559219"}]},"branch":"refs/heads/master"},"f277bb6b4b1f653038d27a08beabf70c0b40b9e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-10 20:30:35.000000000","uploader":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"ref":"refs/changes/67/651567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/2"}}},"commit":{"parents":[{"commit":"d153846e0c2058e21b9161820e3c47b22f381141","subject":"Merge \"Add checking mechanism driver during upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d153846e0c2058e21b9161820e3c47b22f381141"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 20:29:47.000000000","tz":-240},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\nDepends-On: I29d5791acaafa31de15bd58a79302a6b0cb0b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f277bb6b4b1f653038d27a08beabf70c0b40b9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f277bb6b4b1f653038d27a08beabf70c0b40b9e7"}]},"branch":"refs/heads/master"},"615c4446bca70eedf35f577fde31f4b939be2757":{"kind":"REWORK","_number":3,"created":"2019-04-10 23:40:11.000000000","uploader":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"ref":"refs/changes/67/651567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/3"}}},"commit":{"parents":[{"commit":"d153846e0c2058e21b9161820e3c47b22f381141","subject":"Merge \"Add checking mechanism driver during upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d153846e0c2058e21b9161820e3c47b22f381141"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 23:40:06.000000000","tz":-240},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\nDepends-On: I29d5791acaafa31de15bd58a79302a6b0cb0b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/615c4446bca70eedf35f577fde31f4b939be2757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/615c4446bca70eedf35f577fde31f4b939be2757"}]},"branch":"refs/heads/master"},"8559462b52647c420b5175cc9eda915c02e9334e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-04-23 13:22:25.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/67/651567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/4"}}},"commit":{"parents":[{"commit":"d153846e0c2058e21b9161820e3c47b22f381141","subject":"Merge \"Add checking mechanism driver during upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d153846e0c2058e21b9161820e3c47b22f381141"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-04-23 13:22:23.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8559462b52647c420b5175cc9eda915c02e9334e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8559462b52647c420b5175cc9eda915c02e9334e"}]},"branch":"refs/heads/master"},"d6ea91ad18c1a5c306cd2ab00c3d175995849a77":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-23 13:22:28.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/67/651567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/5"}}},"commit":{"parents":[{"commit":"4460b91b0f51e0a2b46b1bcba59a4092b0e3420e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4460b91b0f51e0a2b46b1bcba59a4092b0e3420e"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-04-23 13:22:28.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d6ea91ad18c1a5c306cd2ab00c3d175995849a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d6ea91ad18c1a5c306cd2ab00c3d175995849a77"}]},"branch":"refs/heads/master"},"2d1639129dab5d5262b65169bf725cec0c168c94":{"kind":"REWORK","_number":6,"created":"2019-05-13 13:29:25.000000000","uploader":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"ref":"refs/changes/67/651567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/6"}}},"commit":{"parents":[{"commit":"48ddd368450cdd38c6114efafefeaeb07d4e9651","subject":"Merge \"undercloud: resolve undercloud_*_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/48ddd368450cdd38c6114efafefeaeb07d4e9651"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-05-13 13:29:20.000000000","tz":-240},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\nDepends-On: I29d5791acaafa31de15bd58a79302a6b0cb0b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2d1639129dab5d5262b65169bf725cec0c168c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2d1639129dab5d5262b65169bf725cec0c168c94"}]},"branch":"refs/heads/master"},"1e3e0d98785fe33294ec5214f8fa13c1cdcdf35d":{"kind":"REWORK","_number":7,"created":"2019-05-31 17:51:59.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/7"}}},"commit":{"parents":[{"commit":"2b3b7f1e1f4b7875ff62a6dca8862678f4589c89","subject":"Add a user confirmation to node delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2b3b7f1e1f4b7875ff62a6dca8862678f4589c89"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-31 17:41:45.000000000","tz":-360},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\nDepends-On: I29d5791acaafa31de15bd58a79302a6b0cb0b7dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1e3e0d98785fe33294ec5214f8fa13c1cdcdf35d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1e3e0d98785fe33294ec5214f8fa13c1cdcdf35d"}]},"branch":"refs/heads/master"},"09502e6fadada3ab18a398926fa8dd23214ea125":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-06-03 14:41:27.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/67/651567/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/8"}}},"commit":{"parents":[{"commit":"2b3b7f1e1f4b7875ff62a6dca8862678f4589c89","subject":"Add a user confirmation to node delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2b3b7f1e1f4b7875ff62a6dca8862678f4589c89"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-03 14:41:25.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/09502e6fadada3ab18a398926fa8dd23214ea125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/09502e6fadada3ab18a398926fa8dd23214ea125"}]},"branch":"refs/heads/master"},"5d2f6ae940a97da2ca98b119d72dc17e08e3552f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-06-05 17:09:07.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/67/651567/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/9"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 17:09:07.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d2f6ae940a97da2ca98b119d72dc17e08e3552f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d2f6ae940a97da2ca98b119d72dc17e08e3552f"}]},"branch":"refs/heads/master"},"97cb12003248931c461de3ee18fb1d3d587bb965":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-06-05 20:42:07.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/67/651567/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/10"}}},"commit":{"parents":[{"commit":"42ca45b58e0bde95b87d1dcd03d9aae8b4f4d942","subject":"doc/requirements: require sphinx\u003e\u003d1.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/42ca45b58e0bde95b87d1dcd03d9aae8b4f4d942"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:42:07.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/97cb12003248931c461de3ee18fb1d3d587bb965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/97cb12003248931c461de3ee18fb1d3d587bb965"}]},"branch":"refs/heads/master"},"25934b219aa3ca5e4b6fed14c30222a2804a691f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-06-06 13:24:52.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/11"}}},"commit":{"parents":[{"commit":"5c4907a49d158608a6528726ff992557119fbf3a","subject":"doc/requirements: require sphinx\u003e\u003d1.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5c4907a49d158608a6528726ff992557119fbf3a"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 13:24:52.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/25934b219aa3ca5e4b6fed14c30222a2804a691f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/25934b219aa3ca5e4b6fed14c30222a2804a691f"}]},"branch":"refs/heads/master"},"8c8cde75d005ee8f1c05e2f36e8283ffe935ab0c":{"kind":"NO_CHANGE","_number":12,"created":"2019-06-06 13:25:29.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/12"}}},"commit":{"parents":[{"commit":"cb4eedc597004e9f2caa97c4cc287e9ea2e03a4f","subject":"doc/requirements: fix sphinx requirement lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cb4eedc597004e9f2caa97c4cc287e9ea2e03a4f"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 13:25:29.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8c8cde75d005ee8f1c05e2f36e8283ffe935ab0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8c8cde75d005ee8f1c05e2f36e8283ffe935ab0c"}]},"branch":"refs/heads/master"},"5d2bb7b1c3fcac97e030de24f85da35c63cc3cd7":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-06-06 18:51:41.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/13"}}},"commit":{"parents":[{"commit":"d21648a359ddb0f99cf3606a32930e4e61f3104e","subject":"doc/requirements: fix sphinx requirement lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d21648a359ddb0f99cf3606a32930e4e61f3104e"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 18:51:41.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d2bb7b1c3fcac97e030de24f85da35c63cc3cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5d2bb7b1c3fcac97e030de24f85da35c63cc3cd7"}]},"branch":"refs/heads/master"},"cc9146f18eeb812a0dee546698cf4835ef125beb":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-06-11 13:23:49.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/14"}}},"commit":{"parents":[{"commit":"2bf370cfd00692e28767fa2c383b5110bf280802","subject":"doc/requirements: fix sphinx requirement lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2bf370cfd00692e28767fa2c383b5110bf280802"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-11 13:23:49.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cc9146f18eeb812a0dee546698cf4835ef125beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cc9146f18eeb812a0dee546698cf4835ef125beb"}]},"branch":"refs/heads/master"},"28f6f18cbd77a40fd68fa2a2b1ca71955a0861e4":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-06-12 12:57:01.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/67/651567/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/15"}}},"commit":{"parents":[{"commit":"53d0237c8cee3275e77021a74f661b3df01380a9","subject":"Merge \"Update sphinx requirements for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/53d0237c8cee3275e77021a74f661b3df01380a9"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-12 12:57:01.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/28f6f18cbd77a40fd68fa2a2b1ca71955a0861e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/28f6f18cbd77a40fd68fa2a2b1ca71955a0861e4"}]},"branch":"refs/heads/master"},"50744f7c229096db8cf07199d16485d6a11338f6":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-06-12 15:14:44.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/651567/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/651567/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/651567/16"}}},"commit":{"parents":[{"commit":"6aa1902322e2256b55a40af46224aabfbfd11be7","subject":"doc/requirements: fix sphinx requirement lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6aa1902322e2256b55a40af46224aabfbfd11be7"}]}],"author":{"name":"Dan Prince","email":"dprince@redhat.com","date":"2019-04-10 14:47:34.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-12 15:14:44.000000000","tz":0},"subject":"Wire in convert_docker_params","message":"Wire in convert_docker_params\n\nThis converts Docker*Image params into\nContainer*Image varients when using tripleo deploy.\n\nIn the future we can use it for a few other ad-hoc\nDocker* variable substitutions as well to make things\nmore generic.\n\nChange-Id: Ib1dc0c08ce7971a03639acc42b1e738d93a52f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/50744f7c229096db8cf07199d16485d6a11338f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/50744f7c229096db8cf07199d16485d6a11338f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
