)]}'
{"id":"openstack%2Fpython-tripleoclient~660111","triplet_id":"openstack%2Fpython-tripleoclient~master~I7210831eafea825a3e7f4209c79b07a0b2510dac","project":"openstack/python-tripleoclient","branch":"master","hashtags":[],"change_id":"I7210831eafea825a3e7f4209c79b07a0b2510dac","subject":"Fix exception message","status":"MERGED","created":"2019-05-20 12:57:30.000000000","updated":"2019-05-23 21:21:23.000000000","submitted":"2019-05-23 21:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660111-1558646483875-c3efd882","meta_rev_id":"300e11fda575b329c15a7c9864a9c1ea34b2f596","_number":660111,"virtual_id_number":660111,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-23 21:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-23 17:21:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-05-21 15:47:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-05-22 19:54:00.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-05-22 19:54:00.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 15:47:22.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-22 19:54:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-23 17:21:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-23 21:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e05bb430356c5b6b48ae978a52e34363bac411a","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-05-20 12:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b680e1a8cdf6d3beb4b2bff5d45a4e0de811bbf8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 15:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f3c3cee/ : SUCCESS in 2h 29m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a758d19/ : SUCCESS in 2h 42m 45s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d65f0bb/ : SUCCESS in 1h 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"56b3af683aee0ef6c247cb92ac514b47c389830e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:00:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/11/660111/1/check/openstack-tox-lower-constraints/1130088/ : SUCCESS in 5m 27s\n- openstackclient-check-plugins http://logs.openstack.org/11/660111/1/check/openstackclient-check-plugins/cef5f09/ : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/11/660111/1/check/openstack-tox-pep8/f8c6881/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/11/660111/1/check/openstack-tox-py27/88ffa53/ : SUCCESS in 3m 04s\n- openstack-tox-py35 http://logs.openstack.org/11/660111/1/check/openstack-tox-py35/aa87ef7/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/11/660111/1/check/openstack-tox-py36/ddb8993/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/11/660111/1/check/openstack-tox-py37/544c080/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/11/660111/1/check/openstack-tox-docs/dc447bf/html/ : SUCCESS in 4m 18s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/660111/1/check/tripleo-buildimage-overcloud-full-centos-7/3c819e6/ : SUCCESS in 25m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-containers-multinode/7c5567b/ : SUCCESS in 2h 00m 53s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-undercloud-containers/df6caf7/ : SUCCESS in 1h 53m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/65e0d80/ : SUCCESS in 2h 18m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4ba3155/ : SUCCESS in 2h 35m 57s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone/a488618/ : SUCCESS in 1h 52m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/11/660111/1/check/tripleo-ci-fedora-28-standalone/c97b6e3/ : SUCCESS in 1h 50m 26s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone-upgrade/ec764d3/ : SUCCESS in 1h 58m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/7d7328b/ : SUCCESS in 1h 56m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/536f3f7/ : FAILURE in 1h 42m 06s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/d986ab2/ : SUCCESS in 2h 21m 37s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/00b6664/ : FAILURE in 1h 20m 36s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6e4d209/ : SUCCESS in 2h 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2360fac7e5f8d0e872a3a9cf1363891b5eba87b0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-21 15:47:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e8cf2e1b0511a69f29b5a128a1c0a2790a85778","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-22 19:54:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8330e7b4e35ef7959821b67fda3bfaa9cd608568","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 19:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ddc273bd200b89ae2ec462b4ec2fc90cb5e1849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 21:19:41.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c032e5198a2af4913cd9f53ed20a572f3283d784","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 02:38:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/11/660111/1/gate/openstack-tox-lower-constraints/fbd23d1/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/11/660111/1/gate/openstack-tox-pep8/e0f0cf5/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py27/8e7f10d/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py35/8b32d15/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py36/4b69246/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py37/94f9308/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/11/660111/1/gate/openstack-tox-docs/698a714/html/ : SUCCESS in 3m 33s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/660111/1/gate/tripleo-buildimage-overcloud-full-centos-7/1753cb9/ : SUCCESS in 23m 55s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-containers-multinode/d57897e/ : SUCCESS in 2h 15m 04s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-standalone/002cd9d/ : SUCCESS in 1h 57m 08s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-fedora-28-standalone/2d0fbe0/ : FAILURE in 1h 47m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/854406d/ : SUCCESS in 2h 15m 01s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-standalone-upgrade-stein/dd7f2a6/ : SUCCESS in 2h 11m 22s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-undercloud-containers/43051db/ : SUCCESS in 1h 36m 11s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario001-standalone/8f37607/ : SUCCESS in 1h 47m 27s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario002-standalone/f70fed1/ : SUCCESS in 1h 50m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario003-standalone/0f02088/ : SUCCESS in 1h 19m 47s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario004-standalone/d0221f1/ : SUCCESS in 1h 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"75bfc193b1d90b8038ac99e05ffb00ea952c867e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-23 14:17:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84018940defe03e499d0ddb063c76a1cd81bc973","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 17:21:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b4bbf72/ : FAILURE in 1h 33m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8dfaca1/ : SUCCESS in 2h 50m 25s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/11/660111/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/32f884d/ : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"79cb17e9e007319be221689c80ed93b5009d4fad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 18:45:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/11/660111/1/check/openstack-tox-lower-constraints/80de38a/ : SUCCESS in 5m 11s\n- openstackclient-check-plugins http://logs.openstack.org/11/660111/1/check/openstackclient-check-plugins/9b62e96/ : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/11/660111/1/check/openstack-tox-pep8/48b203c/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/11/660111/1/check/openstack-tox-py27/26fe0a0/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/11/660111/1/check/openstack-tox-py35/499fad5/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/11/660111/1/check/openstack-tox-py36/88f2ee6/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/11/660111/1/check/openstack-tox-py37/6e25195/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/11/660111/1/check/openstack-tox-docs/bab4b57/html/ : SUCCESS in 3m 18s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/660111/1/check/tripleo-buildimage-overcloud-full-centos-7/a8f2019/ : SUCCESS in 23m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-containers-multinode/bac8136/ : SUCCESS in 2h 03m 40s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-undercloud-containers/404beb5/ : SUCCESS in 1h 25m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/25ee745/ : SUCCESS in 2h 07m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/07e4f40/ : SUCCESS in 2h 27m 40s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone/ebd5e9d/ : SUCCESS in 1h 45m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/11/660111/1/check/tripleo-ci-fedora-28-standalone/cdb6aec/ : SUCCESS in 1h 50m 33s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone-upgrade/0b6bd68/ : SUCCESS in 2h 05m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/85315bf/ : SUCCESS in 2h 11m 09s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/674befe/ : FAILURE in 1h 54m 41s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/6ab248b/ : SUCCESS in 1h 57m 33s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/9e0d455/ : FAILURE in 1h 23m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/11/660111/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/ee059d8/ : SUCCESS in 2h 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03aa01059ecbd941588c54cdffaa5a1d81eb9c92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 18:47:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"981cab0699c20efb0352ff444e3120a7cae493a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:21:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"300e11fda575b329c15a7c9864a9c1ea34b2f596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:21:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/11/660111/1/gate/openstack-tox-lower-constraints/c487038/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/11/660111/1/gate/openstack-tox-pep8/fdfeeb6/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py27/6870dcb/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py35/8d3e078/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py36/bd6b56c/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/11/660111/1/gate/openstack-tox-py37/69e1d24/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/11/660111/1/gate/openstack-tox-docs/33c0505/html/ : SUCCESS in 3m 22s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/11/660111/1/gate/tripleo-buildimage-overcloud-full-centos-7/0efe4c9/ : SUCCESS in 24m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-containers-multinode/3948762/ : SUCCESS in 2h 05m 47s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-standalone/f32e23a/ : SUCCESS in 1h 45m 56s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-fedora-28-standalone/f4a7e2a/ : SUCCESS in 1h 57m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/987d6f9/ : SUCCESS in 2h 17m 00s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-standalone-upgrade-stein/715f6ff/ : SUCCESS in 2h 18m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-undercloud-containers/9513767/ : SUCCESS in 1h 58m 29s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario001-standalone/9b424e8/ : SUCCESS in 1h 40m 37s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario002-standalone/aaeab44/ : SUCCESS in 1h 52m 41s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario003-standalone/c704729/ : SUCCESS in 1h 10m 08s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/11/660111/1/gate/tripleo-ci-centos-7-scenario004-standalone/3afdda8/ : SUCCESS in 1h 48m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0989e39c434a62e3a4a3b79e5eb58be96bac514d","revisions":{"0989e39c434a62e3a4a3b79e5eb58be96bac514d":{"kind":"REWORK","_number":1,"created":"2019-05-20 12:57:30.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/11/660111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/11/660111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/660111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/660111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/660111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/11/660111/1"}}},"commit":{"parents":[{"commit":"875bbc74ee03bda871496a45c803dabe32cc2114","subject":"Merge \"Add ODL deprecation warning in CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/875bbc74ee03bda871496a45c803dabe32cc2114"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-05-20 12:56:34.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-05-20 12:56:34.000000000","tz":-240},"subject":"Fix exception message","message":"Fix exception message\n\nFixes wording of the exception message, and also notes that the\ntemplates directory may be inaccessible due to permission denied. This\ncould be caused by using a remote mounted (nfs/sshfs) directory that is\nnot accessible by root.\n\nChange-Id: I7210831eafea825a3e7f4209c79b07a0b2510dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0989e39c434a62e3a4a3b79e5eb58be96bac514d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0989e39c434a62e3a4a3b79e5eb58be96bac514d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
